)]}'
{"id":"openstack%2Fpuppet-keystone~860523","triplet_id":"openstack%2Fpuppet-keystone~master~I6a4e13786e37d25887ddb25c90ed9af3a1d9e899","project":"openstack/puppet-keystone","branch":"master","hashtags":[],"change_id":"I6a4e13786e37d25887ddb25c90ed9af3a1d9e899","subject":"DNM: Testing ...","status":"ABANDONED","created":"2022-10-06 05:32:47.000000000","updated":"2022-10-10 23:28:09.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b1f1a942669f1971a0779c2bd7d95bc3342fa71e","_number":860523,"virtual_id_number":860523,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-10-06 07:42:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-06 07:42:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"904362e135d1f602cfdb5bacab43e4fdb1de83cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-06 05:32:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92eeecbc03b8d19313d823398f4d3023a08f645e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 07:42:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8882660965904675a02e0c367d277573\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/241865046c544dcba2b60131c6f288e2 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/743127cb43b0485cb2fc0286585a5075 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a10b046d4b4f4405a7999e6143050fbe : SUCCESS in 3m 30s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c97929244ac14787a447a83f4a6b8822 : SUCCESS in 18m 49s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/882a8423bed640caa4314e912dd71956 : SUCCESS in 13m 20s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1aef262f15fc4e04b581a16979f4bd79 : SUCCESS in 12m 55s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1fa70db7c124ba7969ecded027924ca : SUCCESS in 12m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/257741238cc74aa38b06a1f14439e8bb : SUCCESS in 52m 12s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc21d5554c7045f79a170c646d15abbf : SUCCESS in 1h 06m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e66e7a7883bf42f99deff1220b01c033 : SUCCESS in 52m 16s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20b639821d9d4a71b7d7acab6cf2267b : SUCCESS in 35m 27s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afe5df4a03a34bb5b0cf151febd52024 : SUCCESS in 43m 41s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8458f5c4d4848b480d3955abaa2abae : SUCCESS in 24m 29s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7981d24808f440c18a9568f1736c7da4 : SUCCESS in 1h 13m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bae068add7ea4c58bb1ce8925a6d0be2 : SUCCESS in 1h 13m 48s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/857ee4855ea0403baabb45bee7d5719b : SUCCESS in 1h 09m 22s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1e27a184f5f44738e630a96fa412313 : SUCCESS in 1h 02m 43s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce4a3a8487cb4eceaf182c457a873d84 : SUCCESS in 1h 01m 11s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27340d2737f640a0a53ef54f15248e2e : SUCCESS in 36m 00s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/bfe0a1620254458d9292115e18ac7f79 : SUCCESS in 43m 59s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb705d86e1ae4174997388c14c3fa698 : SUCCESS in 35m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9e1de3227b54babb8df8fc54fc6264a : SUCCESS in 37m 00s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/18beb526ae174a3ba8109fcc11d31d81 : SUCCESS in 2h 02m 59s","accounts_in_message":[],"_revision_number":1},{"id":"e730307d2cbba077cc0d6978e294b5ece385ba70","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-06 08:17:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36acbf46818f807629bb2929a46cebf2f8927c6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 10:17:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0436d1c0ab93432291add541d1018e8a\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9f61ec0501d40fdb57f889e4d1ecc30 : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fee209a8a1c487ca20358a3c56df99e : SUCCESS in 3m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3a515dd191f457e94832c8be5a78c40 : SUCCESS in 3m 02s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a25d3d98ea745fc89c37c9554fca342 : SUCCESS in 17m 09s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e2be71edc154fe9a59c018ad9f76980 : SUCCESS in 11m 47s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aefb2743e0e54d4891a4d2101f7729dd : SUCCESS in 10m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c061433ea0214b3f84650b41c82baf7a : SUCCESS in 11m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ada33c0bb34a44fd815bed0804a3a3db : SUCCESS in 51m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/904c5ca718194bf0aeeef0de86263498 : SUCCESS in 54m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3feefd1e801a4fc5a9b17dbcc2c7bfdb : SUCCESS in 46m 15s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/414febbde5f242afbfa3bc58b76a1d35 : SUCCESS in 38m 06s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed6e2a2ef476467eb8116c6ea056dcb9 : SUCCESS in 38m 37s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0dfc4d8ce105433bb46ac2e7b754d8ac : SUCCESS in 23m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d29e28007264843a90ae4206cc1331b : SUCCESS in 57m 45s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/07ae8bfcd3d045f1ad7024e738ae16f2 : SUCCESS in 1h 09m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/80e51452cc46443689688aae0ec2367a : SUCCESS in 57m 10s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/665801c5c78d4482b3207ad394053131 : SUCCESS in 49m 57s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1af281e56fbb499f8cb3a982ef60f9d0 : SUCCESS in 53m 35s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b4d98f3a7f540b2bbe7b6d31956947b : SUCCESS in 36m 44s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ad42aa6c15e4950a0d36f00f2be2ca6 : SUCCESS in 36m 21s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d5dd83468fe405bac71b4816db014c3 : SUCCESS in 27m 49s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7723aa6a3331401f92b0402ebedd21f3 : SUCCESS in 39m 24s (non-voting)\n- puppet-keystone-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f9bdb670b8b9499bb237f86a5a839735 : SUCCESS in 1h 53m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b1f1a942669f1971a0779c2bd7d95bc3342fa71e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-10 23:28:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad0a9c290c1605873ac6a832fb3a947ab2d362d8","revisions":{"ad0a9c290c1605873ac6a832fb3a947ab2d362d8":{"kind":"REWORK","_number":1,"created":"2022-10-06 05:32:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/860523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/23/860523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/860523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/860523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/860523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/23/860523/1"}}},"commit":{"parents":[{"commit":"31c42f0dd6cfa659fea5dca224630afd7d0ce426","subject":"Merge \"Allow customizing --date option for trust_flush cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/31c42f0dd6cfa659fea5dca224630afd7d0ce426"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-06 05:32:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-06 05:32:23.000000000","tz":540},"subject":"DNM: Testing ...","message":"DNM: Testing ...\n\nDepends-on: https://review.opendev.org/c/openstack/tripleo-heat-templates/+/860522\nChange-Id: I6a4e13786e37d25887ddb25c90ed9af3a1d9e899\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/ad0a9c290c1605873ac6a832fb3a947ab2d362d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/ad0a9c290c1605873ac6a832fb3a947ab2d362d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
