)]}'
{"id":"openstack%2Fpuppet-keystone~881423","triplet_id":"openstack%2Fpuppet-keystone~master~Ibf470e3ea26075e19d704827c00a73a49976937d","project":"openstack/puppet-keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-28 05:23:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibf470e3ea26075e19d704827c00a73a49976937d","subject":"Refactor keystone::federation::shibboleth","status":"MERGED","created":"2023-04-25 02:52:00.000000000","updated":"2023-04-28 05:23:46.000000000","submitted":"2023-04-28 05:23:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881423","meta_rev_id":"5fd7a06a01268b61a7fc47bb9b2c4ea012a31c9b","_number":881423,"virtual_id_number":881423,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-28 05:23:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-04-27 08:37:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-04-28 04:06:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-04-28 04:06:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-25 04:16:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-27 08:37:34.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-04-28 04:06:07.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"ce5595cbad3ba56e7be6ed991094024b48ed31f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-25 02:52:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"344a04ea087e2ca8d01ffb6190289a19d92fed4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 04:16:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4d02b8182fb41839ab9c192c0475993\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76e00d2bdec14499bfccfc74ffd1f492 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88aa9a3c2569479c846721bea84602be : FAILURE in 3m 16s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d60c188256bf48c38d1e22d26a80317f : FAILURE in 10m 40s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e998dfcf09f473ea7eabb0ca0cf429f : FAILURE in 8m 15s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d75451e003be4d10a67918ebdd2edf67 : FAILURE in 5m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/451a807a14b54b038d3968aafd642760 : SUCCESS in 59m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe3c6fbafccb494884e795c25a1198a2 : SUCCESS in 49m 48s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7a6b0dd6e854cda95e37e37392ce2c0 : SUCCESS in 50m 23s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ee36200f54c43299deffc30e41bf299 : SUCCESS in 38m 56s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21608a3443c7417f9e5bc458b568a3b3 : SUCCESS in 45m 34s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a436e18b8a541bdacabfe1eebfc0d5c : SUCCESS in 17m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/28d71c63b3954c88907e1e923d7eb6cf : SUCCESS in 1h 02m 46s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/693fde6ea2a44ab080cd54382984b915 : SUCCESS in 1h 09m 08s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c215caf23574a5290802f9cb702ea49 : SUCCESS in 1h 04m 48s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7defb99010a04973a11f172eb0164a29 : SUCCESS in 51m 26s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c56eb1302254602b0bc8365d49c136b : SUCCESS in 49m 36s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/0dccc2c5d3dd48ed87679d6a81bcfec1 : FAILURE in 37m 29s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1feebef2d4c44e79801eb2f3d4872d0c : FAILURE in 28m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a64d53fc44434d39ac658ff60dec4b73 : FAILURE in 31m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1b86a4f08ddccb1c8cea5e4cd135450de33c212","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-25 05:01:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"239ba438fda9c4effd0a358cf70312a5f2ca5109","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-25 05:10:11.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"42612f1aa4ca4a1e76c124a1626a8b124d773bcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 06:21:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7f9f03c6b624678ac31d39add72d19a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f270b7fec644b78ae08a53cb3f25279 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f75ffd81c854e55a761eba41b6b4be6 : SUCCESS in 2m 37s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/91510088406e4ba9b26cc5bb7547cf55 : FAILURE in 10m 41s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f0c2dcfbb8b43d9ba37d7ada6dd0655 : FAILURE in 5m 48s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8f83605179745e1a042800b7b33874a : FAILURE in 5m 22s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f1401bc5d3f4514acfde422b4bc49b8 : SUCCESS in 50m 31s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42aeb8c446ca436f9848f632247f2217 : SUCCESS in 1h 04m 44s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd8e6e3622d94b6385986e008a443001 : SUCCESS in 51m 58s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28ea6c88da2d45cebf9b5a8e4e36609f : SUCCESS in 44m 20s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c0ad25d60f64049ad83b4c4a5831163 : SUCCESS in 44m 28s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b450a051f0384de8be6f10392e732b98 : SUCCESS in 19m 27s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/569d5bf5e73e4b318212fad7afd84d63 : FAILURE in 1h 02m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c80b512b01534fcfa901b181bbf0a807 : SUCCESS in 1h 09m 05s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c40ca6f8fa6644f783333a58e36c6c95 : SUCCESS in 1h 08m 16s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/630f2595e17548ff8cf81cf55740410b : SUCCESS in 50m 58s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/da3e8318a56c4cf49086ce52f008de81 : SUCCESS in 51m 38s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/df597908fd6b4844919df469a3982834 : SUCCESS in 38m 19s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c2b35c4e05943d99e135f10539bd243 : SUCCESS in 32m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d0fcf4f44a641bb9e3bf3745d911b09 : SUCCESS in 35m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c43a6767f49c7733d51b9900f542216956f9eed9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-26 13:07:44.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"db51c86a0e6cdadd41713eba8fe90d5d2591d553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 14:44:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e88d21a25c284d6fb5eb7f1486291991\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23a5e5c8b55a4b8d9fe09c3e898e3a5d : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a578f32b55584ecb97df4b866a146d1f : SUCCESS in 2m 22s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff28b96e26e64744a373a0b953809af8 : FAILURE in 12m 12s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11c58c86c40e4b29a910305b4caed89d : FAILURE in 9m 34s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1fd337ace1c54545885efdc603c8f880 : FAILURE in 6m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13ac24de941444deb5e4ea5d0a5e1e4a : SUCCESS in 46m 41s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/841f0af3bc2b446a9eee1b1533df1de0 : SUCCESS in 1h 22m 00s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c72e65a774c64516922c364297415486 : SUCCESS in 50m 10s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/209ad07b2543408d8ac64d3610979ba6 : SUCCESS in 43m 00s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2112b527e9740a9b6be494740ced102 : SUCCESS in 42m 57s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/08d6a207db5e4e0d9ecec498e068593f : SUCCESS in 22m 08s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a60b8611f1dc4eaaabc53520be0d9288 : SUCCESS in 1h 03m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1aa617413a02475782cfd12772c59367 : SUCCESS in 1h 20m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b3268963a434580a98ae3d3a0983bb8 : SUCCESS in 1h 10m 33s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/22352b396df84b8fa9ce8c4a779cc57f : SUCCESS in 55m 54s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/403c714b29254be49bd2bc5268b8d048 : SUCCESS in 58m 27s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed62720b996c4b3e88dcce00766d4319 : SUCCESS in 36m 05s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25beab68ed8f462089619a85a98a2509 : SUCCESS in 35m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06e0e959662f4a009ff6fee9db6c9edf : FAILURE in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"caeb5664aa0ad3c91c55eb7ab64e28c5f0ee7c94","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-27 00:39:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1ae7902453a1631383757b204320a1d377a1b22c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 02:05:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8034823601a748a5a2cb3b977d06279f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db581386cce34853a30d265d05a09c61 : SUCCESS in 3m 08s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d4a915638fa4931ab612e07889ab680 : SUCCESS in 3m 25s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c0b9d63d46c407bacb7558e92a065aa : SUCCESS in 17m 19s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a79f808ad4d9407387c9bbcfc6944359 : SUCCESS in 11m 31s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dba23deb418b4a93a8a3ef77e47d6e11 : SUCCESS in 7m 40s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69acbfa39efa4518ab81372a97aba32e : SUCCESS in 55m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37f87679a62148e29760660fd7e4a061 : FAILURE in 1h 07m 52s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d9c18ad93ca4adab3d18eea50560912 : SUCCESS in 54m 37s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a330779d1d804613a57e37d5110a2285 : SUCCESS in 47m 24s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0bada58f9a34eef81f0e0d1f2704048 : SUCCESS in 35m 08s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4b3ad12d6e7488bad69df52e38acf59 : SUCCESS in 24m 57s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fa2974a26bc479fbc2db23d404e4770 : SUCCESS in 1h 07m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/181b0f2df5404f4a855ed41da2e3a7f9 : SUCCESS in 1h 04m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb8bda6d7ed741ddb5440ec0b9911836 : SUCCESS in 1h 05m 31s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0ebe46585fc46059389a2c247af398d : SUCCESS in 58m 42s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d0e2a52837f4edda7e80745795a84b8 : SUCCESS in 56m 06s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce3b4fdc46db47328849166554d1cff4 : SUCCESS in 31m 53s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcd02f4663764412b984a4c9eca7ee17 : SUCCESS in 32m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b913f49f7b044dbc89898bd979971350 : SUCCESS in 27m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f806149e290e584608ca5477bd61cdae5be250ca","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-04-27 08:37:34.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"22fee3fd40ccb2d531d11896ffef6224b051d0a2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-04-28 04:06:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e3018e4c45baacbff3d9caf24e20b43ad953aded","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 04:07:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0435900e1a8b24f68c236184754e744738058282","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 05:23:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f1b2dabf5264edfb5d7f0bd3b4dc7e0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a619204fc6c4d539ad45f8383efbc07 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb5c8a4351684d3db24a27f08fbb3271 : SUCCESS in 2m 37s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cdfd12393eb4d47ba962985e03cbeef : SUCCESS in 13m 46s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/037e9a7114354c539f295c4ccbf6dec1 : SUCCESS in 9m 42s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/efeae48ab911458fa098614652825a6a : SUCCESS in 51m 03s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/193f4d68ac8a4e69a0b33b910f3e4293 : SUCCESS in 1h 07m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1ce6d983b8240eb8ec32e09bf1c7927 : SUCCESS in 1h 04m 22s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4f5e88cd8a44f95a15bbbcd9ca2d759 : SUCCESS in 55m 50s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/abd6a5979ef94618bcb95574d0d1d976 : SUCCESS in 55m 04s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc720048c0af41cb97fbbea28c0bfe42 : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":4},{"id":"5fd7a06a01268b61a7fc47bb9b2c4ea012a31c9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 05:23:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7f22f87bb66eb99aff34a4a5981c45bc67d457b2","revisions":{"960fcfe2e4a818d3f1f56f55f2a3787bda947b00":{"kind":"REWORK","_number":1,"created":"2023-04-25 02:52:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/881423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/23/881423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/1"}}},"commit":{"parents":[{"commit":"1b756a711c6480b3a2d6172ec9d65c00c65d405c","subject":"Merge \"Bump upper version of puppetlabs-concat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/1b756a711c6480b3a2d6172ec9d65c00c65d405c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-25 02:51:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-25 02:51:10.000000000","tz":540},"subject":"Refactor keystone::federation::shibboleth","message":"Refactor keystone::federation::shibboleth\n\nThis refactors redundant if-else logic of the class to make it more\nreadable. Also this allows us to get rid of the lint failure currently\ndetected in CI.\n\nmanifests/federation/shibboleth.pp:77:WARNING: line has more than 140\ncharacters\n\nChange-Id: Ibf470e3ea26075e19d704827c00a73a49976937d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/960fcfe2e4a818d3f1f56f55f2a3787bda947b00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/960fcfe2e4a818d3f1f56f55f2a3787bda947b00"}]},"branch":"refs/heads/master"},"648bf358266c8ddd42a93cbc69f62f9d5e4c045c":{"kind":"REWORK","_number":2,"created":"2023-04-25 05:01:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/881423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/23/881423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/2"}}},"commit":{"parents":[{"commit":"1b756a711c6480b3a2d6172ec9d65c00c65d405c","subject":"Merge \"Bump upper version of puppetlabs-concat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/1b756a711c6480b3a2d6172ec9d65c00c65d405c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-25 02:51:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-25 05:01:27.000000000","tz":540},"subject":"Refactor keystone::federation::shibboleth","message":"Refactor keystone::federation::shibboleth\n\nThis refactors redundant if-else logic of the class to make it more\nreadable. Also this allows us to get rid of the lint failure currently\ndetected in CI.\n\nmanifests/federation/shibboleth.pp:77:WARNING: line has more than 140\ncharacters\n\nChange-Id: Ibf470e3ea26075e19d704827c00a73a49976937d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/648bf358266c8ddd42a93cbc69f62f9d5e4c045c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/648bf358266c8ddd42a93cbc69f62f9d5e4c045c"}]},"branch":"refs/heads/master"},"6c7bfa6f43d8ef7e5bb34e6708402126669dc339":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-04-25 05:10:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/881423/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/23/881423/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/3"}}},"commit":{"parents":[{"commit":"1b756a711c6480b3a2d6172ec9d65c00c65d405c","subject":"Merge \"Bump upper version of puppetlabs-concat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/1b756a711c6480b3a2d6172ec9d65c00c65d405c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-25 02:51:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-25 05:10:11.000000000","tz":0},"subject":"Refactor keystone::federation::shibboleth","message":"Refactor keystone::federation::shibboleth\n\nThis refactors redundant if-else logic of the class to make it more\nreadable. Also this allows us to get rid of the lint failure currently\ndetected in CI.\n\nmanifests/federation/shibboleth.pp:77:WARNING: line has more than 140\ncharacters\n\nDepends-on: https://review.opendev.org/881425\nChange-Id: Ibf470e3ea26075e19d704827c00a73a49976937d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/6c7bfa6f43d8ef7e5bb34e6708402126669dc339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/6c7bfa6f43d8ef7e5bb34e6708402126669dc339"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7f22f87bb66eb99aff34a4a5981c45bc67d457b2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-04-26 13:07:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/881423/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/23/881423/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/23/881423/4"}}},"commit":{"parents":[{"commit":"1b756a711c6480b3a2d6172ec9d65c00c65d405c","subject":"Merge \"Bump upper version of puppetlabs-concat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/1b756a711c6480b3a2d6172ec9d65c00c65d405c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-25 02:51:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-26 13:07:44.000000000","tz":0},"subject":"Refactor keystone::federation::shibboleth","message":"Refactor keystone::federation::shibboleth\n\nThis refactors redundant if-else logic of the class to make it more\nreadable. Also this allows us to get rid of the lint failure currently\ndetected in CI.\n\nmanifests/federation/shibboleth.pp:77:WARNING: line has more than 140\ncharacters\n\nChange-Id: Ibf470e3ea26075e19d704827c00a73a49976937d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/7f22f87bb66eb99aff34a4a5981c45bc67d457b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/7f22f87bb66eb99aff34a4a5981c45bc67d457b2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
