)]}'
{"id":"openstack%2Fpuppet-keystone~900135","triplet_id":"openstack%2Fpuppet-keystone~master~I3d4fbb082cf228ef4a75c0761fb21fdebf664cf4","project":"openstack/puppet-keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-07 21:06:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3d4fbb082cf228ef4a75c0761fb21fdebf664cf4","subject":"Use openstack cli to resolve project/user id","status":"MERGED","created":"2023-11-05 15:57:50.000000000","updated":"2023-11-07 21:06:33.000000000","submitted":"2023-11-07 21:06:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900135","meta_rev_id":"f0b292160da487308df31f890b7322900bd173aa","_number":900135,"virtual_id_number":900135,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-07 21:06:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-11-06 09:10:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-11-07 19:32:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-11-07 19:32:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-05 18:01:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-06 09:10:53.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-11-07 19:32:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"6183a8781f0d718b2edbf2e537f30d51ccfe82c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-05 15:57:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"294f875f79229cab4c999390b20000a3b180598f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-05 15:59:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"13afa9ccae7a41a701040a3d580e16ede094f360","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-05 16:21:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fa3408161f0f34d8d96b70aca714a51d12e33f11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-05 16:37:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3d1c5d5bff40d72128a6d32c5096f7aeb4814fa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-05 18:01:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14cff3a9a66b41478e2cc96f14b4e92e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5face8499b94f23a3737bddece64845 : SUCCESS in 2m 20s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c2ac68007d24f3aaf832b16315ccf6d : SUCCESS in 2m 26s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/078930d4ecd54b66b8cb1853288b7f85 : SUCCESS in 4m 42s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ec57d5d736d4917a7392dff6393ce8e : SUCCESS in 14m 00s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abef32de60c3435e9b5264a55cef2bab : SUCCESS in 11m 12s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/47ddecc90af546daa4a1128fee9e02b2 : SUCCESS in 10m 07s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad27b4c74a724e22a12f18201dda1751 : SUCCESS in 11m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3a11fad4ade482dbb7fbb2a9326ca12 : SUCCESS in 37m 18s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcbe4dec925042e185df8dfc1cf5f35e : SUCCESS in 58m 13s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61aac5ba42794443bba12bc8aa89ffed : SUCCESS in 48m 03s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac6384ba1bce4f119a6d571e7be93731 : SUCCESS in 33m 50s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7158293285b4002b2cd2678aedeab34 : SUCCESS in 41m 06s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba9ee609ab6141419074f43b066944c8 : SUCCESS in 45m 27s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d755807274b4f328cf1f01be52e1740 : SUCCESS in 1h 17m 24s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4f8e22e0f884bb99e9e17e6f3b319a9 : SUCCESS in 46m 51s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6c645ac97894e72aff9849e19662d60 : SUCCESS in 57m 25s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ccb7bfc3d7a34ea585fd6284bdf19abf : SUCCESS in 48m 59s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/08f468d1c30948c29975fffcb293867c : SUCCESS in 28m 01s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8662c40ebdd4266b0fcef01342c81f0 : FAILURE in 42m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/adee89ce921640259430376b2e3f8cd0 : FAILURE in 38m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bd6cbd2aa61d6313359cc0897740a9871bc3e7c2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-11-06 09:10:53.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9dffae9eb377b844988fc2f39c8941fdb330f86e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-11-07 19:32:30.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4a217398135d72ecca2a14504d1bd73e3f69adb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 19:33:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bc0e359f342ddde64c011dd4951bc2b5c85aabce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 21:06:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14fa2fca4b144285bf007aff2a345536\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/275026bf1ce34799b492aa92423f0109 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17cb21eda06f443b84e5b6b7f90b6e25 : SUCCESS in 2m 49s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/914b608346a94495b5459b6c9e14dea5 : SUCCESS in 14m 35s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1371fce08ead4f67940f06ef94ec4b65 : SUCCESS in 9m 20s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/df769c58d3484828b9310b9710fb7c10 : SUCCESS in 13m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c0e5de6463c4de0957c59c1d93a613d : SUCCESS in 1h 05m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b002dc7ffb984221b03c7697de3b2b94 : SUCCESS in 1h 28m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ef50f7a41da4d74bdb78d52216083e1 : SUCCESS in 1h 11m 02s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bec8fe87bc7842e8bdfededc91178610 : SUCCESS in 1h 01m 25s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/78adb6f1fd384be6a33d7aaacf328f00 : SUCCESS in 56m 19s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/b482b653edc14b5084fbae5fb01612f5 : SUCCESS in 38m 54s","accounts_in_message":[],"_revision_number":4},{"id":"f0b292160da487308df31f890b7322900bd173aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 21:06:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"80a1953d7d7780c0ec1f6650401ca41e6d504f49","revisions":{"7aded672c5f1a994b09b9cd72e192dc2fc763633":{"kind":"REWORK","_number":1,"created":"2023-11-05 15:57:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/900135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/35/900135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/1"}}},"commit":{"parents":[{"commit":"df9ce566c6a9f8f7390faea9f72ba56eaf613142","subject":"Merge \"Bump upper version of puppetlabs-apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/df9ce566c6a9f8f7390faea9f72ba56eaf613142"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 15:56:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 15:56:53.000000000","tz":540},"subject":"Use openstack cli to resolve project/user id","message":"Use openstack cli to resolve project/user id\n\nThe openstack command can resolve project id or user id from name and\ndomain name/id given. We can use that feature instead of maintaining\nour own logic.\n\nChange-Id: I3d4fbb082cf228ef4a75c0761fb21fdebf664cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/7aded672c5f1a994b09b9cd72e192dc2fc763633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/7aded672c5f1a994b09b9cd72e192dc2fc763633"}]},"branch":"refs/heads/master"},"6afdd461fe10c473cd2899b4110e288bd5da7aea":{"kind":"REWORK","_number":2,"created":"2023-11-05 15:59:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/900135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/35/900135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/2"}}},"commit":{"parents":[{"commit":"df9ce566c6a9f8f7390faea9f72ba56eaf613142","subject":"Merge \"Bump upper version of puppetlabs-apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/df9ce566c6a9f8f7390faea9f72ba56eaf613142"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 15:56:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 15:59:07.000000000","tz":540},"subject":"Use openstack cli to resolve project/user id","message":"Use openstack cli to resolve project/user id\n\nThe openstack command can resolve project id or user id from name and\ndomain name/id given. We can use that feature instead of maintaining\nour own logic.\n\nChange-Id: I3d4fbb082cf228ef4a75c0761fb21fdebf664cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/6afdd461fe10c473cd2899b4110e288bd5da7aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/6afdd461fe10c473cd2899b4110e288bd5da7aea"}]},"branch":"refs/heads/master"},"3747684b31c0e22c58d485f4442ff4a295aa20ee":{"kind":"REWORK","_number":3,"created":"2023-11-05 16:21:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/900135/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/35/900135/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/3"}}},"commit":{"parents":[{"commit":"df9ce566c6a9f8f7390faea9f72ba56eaf613142","subject":"Merge \"Bump upper version of puppetlabs-apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/df9ce566c6a9f8f7390faea9f72ba56eaf613142"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 15:56:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 16:21:13.000000000","tz":540},"subject":"Use openstack cli to resolve project/user id","message":"Use openstack cli to resolve project/user id\n\nThe openstack command can resolve project id or user id from name and\ndomain name/id given. We can use that feature instead of maintaining\nour own logic.\n\nChange-Id: I3d4fbb082cf228ef4a75c0761fb21fdebf664cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/3747684b31c0e22c58d485f4442ff4a295aa20ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/3747684b31c0e22c58d485f4442ff4a295aa20ee"}]},"branch":"refs/heads/master"},"80a1953d7d7780c0ec1f6650401ca41e6d504f49":{"kind":"REWORK","_number":4,"created":"2023-11-05 16:37:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/900135/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/35/900135/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/35/900135/4"}}},"commit":{"parents":[{"commit":"df9ce566c6a9f8f7390faea9f72ba56eaf613142","subject":"Merge \"Bump upper version of puppetlabs-apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/df9ce566c6a9f8f7390faea9f72ba56eaf613142"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 15:56:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 16:37:17.000000000","tz":540},"subject":"Use openstack cli to resolve project/user id","message":"Use openstack cli to resolve project/user id\n\nThe openstack command can resolve project id or user id from name and\ndomain name/id given. We can use that feature instead of maintaining\nour own logic.\n\nChange-Id: I3d4fbb082cf228ef4a75c0761fb21fdebf664cf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/80a1953d7d7780c0ec1f6650401ca41e6d504f49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/80a1953d7d7780c0ec1f6650401ca41e6d504f49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
