)]}'
{"id":"openstack%2Fpuppet-keystone~912824","triplet_id":"openstack%2Fpuppet-keystone~master~I5c55a06b39118156db845c4ee6d415cb4d695ea0","project":"openstack/puppet-keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-14 19:45:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c55a06b39118156db845c4ee6d415cb4d695ea0","subject":"Deprecate amqp messaging driver support","status":"MERGED","created":"2024-03-14 00:36:21.000000000","updated":"2024-03-14 19:46:35.000000000","submitted":"2024-03-14 19:45:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912824","meta_rev_id":"d8c76ed55851bb6dd8bd1001656fb4bef226b1b9","_number":912824,"virtual_id_number":912824,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-14 19:45:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-03-14 11:03:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-03-14 16:55:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-03-14 16:55:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-14 01:42:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-14 11:03:34.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-03-14 16:55:35.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"07a1bcfbe1af1fd2931add6620a802f84f338081","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-14 00:36:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ae3e2a3fcc948dafd169c49adfdd87232bc9519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 01:42:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f7b23851d3e4ef0b82901b8818a4a08\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ec61335e987417281ff87769d995d2f : SUCCESS in 2m 23s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6af7116faf8c48bd9cc07416e1b13a2e : SUCCESS in 2m 20s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfe25d13887c491b8703100d2dcebfaf : SUCCESS in 5m 31s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/085c7ffe86b0426b8de7b413909a7a50 : SUCCESS in 12m 53s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5924af7f3f514a9e977b5594565b7b46 : SUCCESS in 9m 21s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/45646c67dda34faaa631444a43e71d43 : SUCCESS in 13m 03s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b93c5efe0d6447668f5eb1bab3433683 : SUCCESS in 9m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0328ca36ee264058ab73112c98b5b01b : SUCCESS in 37m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0cd83cb234e14a07b77c795498fed90f : SUCCESS in 57m 19s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efcfb2d2faa4496d962dd27df134f7c5 : SUCCESS in 41m 09s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e11cb278db44e9e96f3699756daca7f : SUCCESS in 35m 49s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a1dd6a3088b4d96ad877b33eb6dcebf : SUCCESS in 44m 14s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/63c3ebd38a7f4b3fb0c0d9b3f86d2e96 : SUCCESS in 51m 54s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab858e95a1e74773a91184c5ee64636b : FAILURE in 1h 01m 11s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d0aa54a81e54ba9ada00ac8f012bc54 : SUCCESS in 57m 12s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/55bcd3db2c3543639fc19e33f150bae5 : SUCCESS in 55m 42s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1c98485ac3f45ad81ac8140af81162b : SUCCESS in 54m 08s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f5880f8af0a44139dbe0e2a4228f9f8 : SUCCESS in 37m 27s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bada39c4bce84983a32344bb52cf9030 : SUCCESS in 32m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1bea6a038f04ccc9bd81409e2020d22 : SUCCESS in 30m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e313b1aa9ae4994b65dee28fc2810dd : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fbccfe33e733aaa092594fea6a33e52c0670c02f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-14 02:01:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7563c61b370b0d3e93cec84edaab3886efbb711b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 04:21:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5324b689ec9a4d9fbc3eec0d4160268a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4bfca47f5d0f4fac8af2b42403cb2364 : SUCCESS in 5m 03s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/869393606b5d4a4dae20afcf284443df : SUCCESS in 3m 31s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/9142c9e5ac454c95bd1defc7e993e69f : SUCCESS in 4m 51s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7615db5f961428fac91ece42d2b7327 : SUCCESS in 11m 33s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/267963f7527d4679965b441737fdc33f : SUCCESS in 11m 30s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb869e4308704577a608277af0f35794 : SUCCESS in 14m 37s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a4ef307cfb14070adc111a3bbf81a51 : SUCCESS in 9m 14s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82daf631d53f4c51831e760bbcb43a6d : SUCCESS in 43m 40s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f74e846c83544d24b2938deff23337e5 : SUCCESS in 1h 01m 41s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d41336d9348c4966baa961fb07de7752 : SUCCESS in 51m 57s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32930418ff954fe58f2519fbcd1d7590 : SUCCESS in 41m 14s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34c731c2c39445089eda696fd22d3786 : SUCCESS in 41m 45s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4a8168b7d03415794c6e39cfe2131cf : SUCCESS in 57m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bbae32f5371c4273bb7183535960389d : FAILURE in 1h 25m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/68335745b84049688601f03b7530c57c : SUCCESS in 1h 02m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/54f28683200c4036be18f339b17b206d : SUCCESS in 1h 00m 31s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7aa51ecaafe5437cac0a9403d04b922d : SUCCESS in 1h 04m 10s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c823b2cd6704d8bb43a68015a5368dc : SUCCESS in 39m 44s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c6ddc4912ea401883e11b23812bb752 : SUCCESS in 33m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ede9c432cde46cf875596cda5ea3a52 : SUCCESS in 33m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e370f45f7af849e7930ed3ae1131d096 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"38a8c4e2d8c7ade8d099195b3c1eb194acc19e6b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-14 04:46:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1864d3691912b7a007aa21006f8b4533ee73839e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 06:15:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/380e4fe350c8462396540be634239566\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a89f6ca8e1ad4c90af2109de600085f7 : SUCCESS in 2m 28s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69c37065c8364f1bac79487768b4e968 : SUCCESS in 2m 12s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/846ced9e609d4a17998798414c957978 : SUCCESS in 4m 40s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0cecea2d7f34b0e8dcdc72b5622ebc2 : SUCCESS in 11m 53s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0aa4ae47c515448aaddde56d1e2861e6 : SUCCESS in 9m 26s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/dccbae0730684400b1f3e91ceedf851b : SUCCESS in 14m 08s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4ab3142e3bc4273962a297d6de5df76 : SUCCESS in 8m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9233dfa095a4b45b19d8f378a5b409a : SUCCESS in 44m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4203ce8a5d194545b6272325f5289379 : SUCCESS in 1h 00m 38s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc92bc4821ea4d1299b801cafea52087 : SUCCESS in 41m 16s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c842d7acf4bd44f383a6339b47aaac2b : SUCCESS in 42m 42s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08d5c96b2c5c407bafc739c67833e49a : SUCCESS in 36m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cf5bbf3533e408db3f6f3db25cacff8 : SUCCESS in 57m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/553a1517c8af40b2a7b15dde9812fb2b : SUCCESS in 1h 23m 34s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae17aeaa3ae14e62a7f17869a941d961 : SUCCESS in 47m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a2f06f71c56449dad9ef2afde72842a : SUCCESS in 56m 02s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3b8735e062c499b917deb60a3a219a8 : SUCCESS in 1h 08m 19s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/66fec0aef03343b0b005b74c65121856 : SUCCESS in 40m 27s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a960c823eb8422cb969f05b2d708e2d : SUCCESS in 31m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/820f6613908044ecaa3a3c1538a75098 : SUCCESS in 32m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd54e57d18ad41fca498ffbc078fbb04 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"69c19cdbed1581ae31b51d9d6a379e4ed3feabe3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-03-14 11:03:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2bad826cd21814360dcd3ae3b3db9d0a3f9ac14b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-03-14 16:55:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1f59e2507d0e671dbec8c630c198bc4a515614f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 16:56:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"adefea80d863136ac49044e22e83ad2d42e84fb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 19:45:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edbb034b583e427a81f1d5ede0e05852\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e4d6dc933424627be1f88a16062717a : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef250fcb828d4c77a9946539200b4990 : SUCCESS in 2m 36s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/273db782b6f34532a97ff0e2683008d8 : SUCCESS in 11m 01s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d831d7fe74444a58f39542324e2684b : SUCCESS in 7m 55s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f07efb15d8634a48b4c9d0f3bf954067 : SUCCESS in 1h 00m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/69422f76b7bf432a91b8ceff7381e240 : SUCCESS in 1h 21m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/49600fe1a9d54d38a29a67e637ff3a9e : SUCCESS in 1h 02m 22s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/98f3ec4d10a443faa1c66d5828bec016 : SUCCESS in 1h 01m 57s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/124dc266f21347cca3ab058773eb2db7 : SUCCESS in 51m 19s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/d41139e551aa4c599439329b6743b8a7 : SUCCESS in 38m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f39c45699414d9b837914b10f2bc11e : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b9b6a38dba3c176cf56207f53d8fc4a4149d54d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 19:45:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d8c76ed55851bb6dd8bd1001656fb4bef226b1b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 19:46:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdadea0dfa744221b0b27d9d4706fd58\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13cb7b6da0cb40929d13b24101d6876c : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e79239531eb6b11756dd188887021c8fe81531f","revisions":{"5e79239531eb6b11756dd188887021c8fe81531f":{"kind":"REWORK","_number":1,"created":"2024-03-14 00:36:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/912824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/24/912824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/24/912824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/24/912824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/24/912824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/24/912824/1"}}},"commit":{"parents":[{"commit":"50b5260cc67135b95093be1d4e073584675a1aa2","subject":"Merge \"openidc: Support more redis cache options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/50b5260cc67135b95093be1d4e073584675a1aa2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-14 00:35:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-14 00:35:53.000000000","tz":540},"subject":"Deprecate amqp messaging driver support","message":"Deprecate amqp messaging driver support\n\n... because the driver has been deprecated in oslo.messaging.\n\nChange-Id: I5c55a06b39118156db845c4ee6d415cb4d695ea0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5e79239531eb6b11756dd188887021c8fe81531f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/5e79239531eb6b11756dd188887021c8fe81531f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
