)]}'
{"id":"openstack%2Fpuppet-keystone~917925","triplet_id":"openstack%2Fpuppet-keystone~master~I88ac699a9303deee2f12fed1b257eb72abfb3305","project":"openstack/puppet-keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-10 13:37:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I88ac699a9303deee2f12fed1b257eb72abfb3305","subject":"Remove deprecated and ineffective client_package_ensure","status":"ABANDONED","created":"2024-05-02 07:58:24.000000000","updated":"2025-01-10 13:37:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6bbc623aea388453d83ee901c3d4c3a4bac14a7e","_number":917925,"virtual_id_number":917925,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-22 04:41:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-08-22 09:26:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-02 09:40:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-22 09:26:42.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"459a6116673bc8101895b8a7282596034502191c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-02 07:58:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e062fa68a113be9ff784cd8f1be1f1cf0ef4d824","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-02 08:05:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"db75991e0498b592ef52a486a66870824032e9df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 09:40:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dada06cf94ef4439b5950db8dfd3a9e7\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a66cbdb1dd7748d693fafd0377d86580 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59049299a8ac4195883e386abd0da034 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/310d7dd0e6924e288bc86b397a146a92 : SUCCESS in 5m 24s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e50a9e74a9244df9ab8d556b3e89a28 : SUCCESS in 10m 55s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9151336ab8f14c898f6c3202942ec282 : SUCCESS in 9m 44s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/8689dd1d3f594a1eb26ceb4ff573e064 : SUCCESS in 12m 55s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7140a6550a9b4d48811919b8d689aabc : SUCCESS in 10m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cc897286fa14caa8ab2b919bbd26a45 : SUCCESS in 41m 01s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94f7ce2a54e7403cba1b73402f2b08f6 : SUCCESS in 55m 45s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5eded738481f493fb8efea01bc102d86 : FAILURE in 49m 36s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82633be640bc465bb75c13dfcc65cd79 : SUCCESS in 43m 36s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28f761d974ab41dba1322b1caa5d617b : SUCCESS in 37m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ae50b1d89d34f6e8eee1ea60dbc18a2 : SUCCESS in 52m 38s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9471680023fe46379a332e3afaf25481 : FAILURE in 1h 06m 54s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9c158d0ea4047f5a4fac0af749b7c61 : SUCCESS in 51m 37s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/33c00df7279c4efdb71170ac1be02025 : SUCCESS in 1h 00m 20s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/417e694b4df5459ca9ce85df70d2f3b0 : SUCCESS in 59m 01s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/4921efab5ae8403a91812211e2610ce5 : SUCCESS in 37m 10s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c6777a50a6740e1a6477b46db863ed1 : SUCCESS in 31m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9d519f85a504eba927916daee8d0cca : SUCCESS in 35m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2113ac0d880414ba3625c288e1f9e77 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"1cc9da5526ce50b48a58a008a72569dd44bd3996","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-05 02:26:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f163060f90745a119b74a053ffd87c5f8b693357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 04:25:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08fddb2a54824731b987bf86608331a5\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ce84e4b547b4ba39ce197a81d782197 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f0b28a6a81440b593c126cfe6e30f54 : SUCCESS in 3m 37s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd35a531477c42c087fdaf1b26d117bb : SUCCESS in 5m 39s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7dc38c79e17468ba4b5278c184aaf09 : SUCCESS in 13m 40s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4f25084d5d040b5a6f13b999d983038 : SUCCESS in 8m 32s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/582c918c549a4a4c92f684af6a51fbc5 : SUCCESS in 14m 55s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efc8bc563a824c998f787affd382d4c8 : SUCCESS in 10m 20s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f30ff81a25904e08894c5f09762beb5d : SUCCESS in 43m 41s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ee993856dc4473992037b82df5bbdaa : SUCCESS in 57m 01s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce54fe60206e4a4c9de96f91d55e601c : FAILURE in 45m 20s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b7d276421ab4bfa80a2dd552f9a2e90 : SUCCESS in 43m 57s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d224b77dd8be4afcbd8b0e28af020965 : SUCCESS in 47m 03s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/48534abc95974fafab319a2d78cd5cd0 : SUCCESS in 1h 00m 08s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/77870b5610474133a776f466a26ada23 : SUCCESS in 1h 41m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f757943ece95411a996b4a266036f80e : SUCCESS in 1h 05m 59s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/738ab7d95a1c4692a7d6cf99a854c919 : SUCCESS in 1h 02m 15s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/756572b091cf49f1bbc7e8058663c28c : SUCCESS in 1h 01m 49s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c4b63f05424430586ea7d01c4390604 : SUCCESS in 34m 54s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0397f3cfff9647efbeff7d1b46cadf72 : SUCCESS in 29m 56s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c03bd2bf86a4678bcfe54ec3b0b3383 : SUCCESS in 43m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b833474516a4daab7eb80778547d03d : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":3},{"id":"b123ef00547ed76259d1fee3f16189dd2b90f190","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-22 03:29:44.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"3d73f9df78117d335f1fb9a1eb7dd2963bebece9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 04:41:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f12a2c85f7b429a9a4e82be12c7bc45\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e67716045d9407d8e1f311a661f9447 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/248c197680114aad847966417828cc80 : SUCCESS in 2m 21s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/2de914d6a288456ca5cfa8e43b431d8a : SUCCESS in 5m 31s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/2705165d255f4557ae0d0aaeecf91a9b : SUCCESS in 11m 32s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fa3ba49957d4e579bc4a727b25c7185 : SUCCESS in 11m 02s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4d0ee716f6c42159aa661e4376ebd19 : SUCCESS in 12m 35s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30a4a9b41bd645ae90e5e65b3e2a32bc : SUCCESS in 10m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2e9dcfddc1543689d3f8e1a5ee55944 : SUCCESS in 40m 25s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d4686119ad64fff8494daf626784156 : SUCCESS in 1h 00m 54s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8476383810fd44ee9b99b682991fbbc2 : SUCCESS in 48m 42s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cff07b85af154c2b8e9a2739f5dee438 : SUCCESS in 44m 34s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a52e62036ddc4899a0248ae615b33550 : SUCCESS in 38m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1db196f05c54e6d999e759a01ac27db : SUCCESS in 56m 43s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6052a0963334572b1079e1536dd1b8d : SUCCESS in 1h 05m 38s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/54f33d37fd8943258b0494f9d6650bb7 : SUCCESS in 54m 51s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/17ce5a7f32824a17b43839147f240731 : SUCCESS in 56m 15s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/65c80cc408724bfa86fbf57132283b5c : SUCCESS in 49m 25s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b9291849cb54207aae042a7925557ab : SUCCESS in 32m 54s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb18a96a228a431aa129fc9e376f1554 : SUCCESS in 26m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a31f7ddf8a9a4863a8c4596f31039b07 : SUCCESS in 30m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b55134b18c7d432bb6a8c395add9ff1b : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":4},{"id":"c945d287bb779ea38cc47593850cf57527de986c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-08-22 09:26:42.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6bbc623aea388453d83ee901c3d4c3a4bac14a7e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-10 13:37:22.000000000","message":"Abandoned\n\nThis was done by 49eaed8a2ca17e3c9a63b3aca80d42787f65ad05","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3478f50de0376a163d20ffd50658a9ac50e78aa3","revisions":{"4267efa4b3bc2d5595f7b13792cfe22cde0850ae":{"kind":"REWORK","_number":1,"created":"2024-05-02 07:58:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/917925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/25/917925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/1"}}},"commit":{"parents":[{"commit":"406344794141a09240e4d579cf0b5fb9f582bf23","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/406344794141a09240e4d579cf0b5fb9f582bf23"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 07:57:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 07:57:49.000000000","tz":540},"subject":"Remove deprecated and ineffective client_package_ensure","message":"Remove deprecated and ineffective client_package_ensure\n\nThe parameter was deprecated in the past cycle[1] and has had no effect\nsince then.\n\n[1] f271472b4824cdad1d0939c7c2d3a68556d1467d\n\nChange-Id: I88ac699a9303deee2f12fed1b257eb72abfb3305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/4267efa4b3bc2d5595f7b13792cfe22cde0850ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/4267efa4b3bc2d5595f7b13792cfe22cde0850ae"}]},"branch":"refs/heads/master"},"3857257f769a8f65ee54a1467cb0c7331ea8aa56":{"kind":"REWORK","_number":2,"created":"2024-05-02 08:05:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/917925/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/25/917925/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/2"}}},"commit":{"parents":[{"commit":"406344794141a09240e4d579cf0b5fb9f582bf23","subject":"Update master for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/406344794141a09240e4d579cf0b5fb9f582bf23"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 07:57:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 08:04:57.000000000","tz":540},"subject":"Remove deprecated and ineffective client_package_ensure","message":"Remove deprecated and ineffective client_package_ensure\n\nThe parameter was deprecated in the past cycle[1] and has had no effect\nsince then.\n\n[1] f271472b4824cdad1d0939c7c2d3a68556d1467d\n\nChange-Id: I88ac699a9303deee2f12fed1b257eb72abfb3305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/3857257f769a8f65ee54a1467cb0c7331ea8aa56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/3857257f769a8f65ee54a1467cb0c7331ea8aa56"}]},"branch":"refs/heads/master"},"0c06e5847a49e3adf75b174a35c4fbf5675bb508":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-07-05 02:26:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/917925/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/25/917925/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/3"}}},"commit":{"parents":[{"commit":"68e091978887937a46bf9dd389e4b9d85079ac59","subject":"feat: add a param for arbitrary federation OIDCRedirectURI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/68e091978887937a46bf9dd389e4b9d85079ac59"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 07:57:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-05 02:26:50.000000000","tz":0},"subject":"Remove deprecated and ineffective client_package_ensure","message":"Remove deprecated and ineffective client_package_ensure\n\nThe parameter was deprecated in the past cycle[1] and has had no effect\nsince then.\n\n[1] f271472b4824cdad1d0939c7c2d3a68556d1467d\n\nChange-Id: I88ac699a9303deee2f12fed1b257eb72abfb3305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/0c06e5847a49e3adf75b174a35c4fbf5675bb508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/0c06e5847a49e3adf75b174a35c4fbf5675bb508"}]},"branch":"refs/heads/master","description":"Rebase"},"3478f50de0376a163d20ffd50658a9ac50e78aa3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-08-22 03:29:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/917925/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/25/917925/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/25/917925/4"}}},"commit":{"parents":[{"commit":"c3c7c39e1c7e0b39c5cc78d4e6efaf7de2904248","subject":"Add database_slave_connection to keystone::db class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/c3c7c39e1c7e0b39c5cc78d4e6efaf7de2904248"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-02 07:57:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-22 03:29:44.000000000","tz":0},"subject":"Remove deprecated and ineffective client_package_ensure","message":"Remove deprecated and ineffective client_package_ensure\n\nThe parameter was deprecated in the past cycle[1] and has had no effect\nsince then.\n\n[1] f271472b4824cdad1d0939c7c2d3a68556d1467d\n\nChange-Id: I88ac699a9303deee2f12fed1b257eb72abfb3305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/3478f50de0376a163d20ffd50658a9ac50e78aa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/3478f50de0376a163d20ffd50658a9ac50e78aa3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
