)]}'
{"id":"openstack%2Fpuppet-keystone~918495","triplet_id":"openstack%2Fpuppet-keystone~master~I0b5b54540242027707d94409be67fa1d73d037ea","project":"openstack/puppet-keystone","branch":"master","topic":"remove-amqp1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-09 15:40:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0b5b54540242027707d94409be67fa1d73d037ea","subject":"Remove deprecated amqp1 messaging driver support","status":"MERGED","created":"2024-05-08 03:23:14.000000000","updated":"2024-05-09 15:41:36.000000000","submitted":"2024-05-09 15:40:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918495-remove-amqp1","meta_rev_id":"c095ac772d9789cd059619cd363b4104a7079a01","_number":918495,"virtual_id_number":918495,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-09 15:40:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-05-08 11:55:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-05-09 14:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-05-09 14:22:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-08 04:38:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-08 11:55:01.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-05-09 14:22:17.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"31d1d202f3b066537aa6269f2edea37345af31df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-08 03:23:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a07428c13790098a6ce25537681f90704dab1ffe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-08 03:28:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"767dfde25be26927c998dc06382fb24e2105c9ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-08 03:29:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d26d9a0d5d5975e3ec90e718f8876faeaf0c2850","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-08 03:30:35.000000000","message":"Topic remove-ampq1 removed","accounts_in_message":[],"_revision_number":3},{"id":"42b1c5a40d843faa29cd7a94d136996629e657ec","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-08 03:30:40.000000000","message":"Topic set to remove-amqp1","accounts_in_message":[],"_revision_number":3},{"id":"dce432f2f957985716f28bc104b3dbc5f2c0a22b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 04:38:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90c8577f4fdb4d1cabdc6bf2ef799722\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e6ae6efb1cf41ca8624d1429a5a3e22 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4110e1ae0a544dc9a6e2831d400d2a2 : SUCCESS in 2m 30s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ef1db7b6c4a46eba9ff9b0864dfbdbe : SUCCESS in 5m 25s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/192a801368b14172b7f6b582c2297f7f : FAILURE in 12m 55s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4620f03d15c4337a603373002bdc562 : FAILURE in 12m 36s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1fedf0c9e6e4009bb808166ff3bdbc0 : FAILURE in 13m 44s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc031d7f99a74bfd9aec5e085890bef9 : FAILURE in 9m 27s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56d4bb4eb55442848b66e7b8dfc50960 : SUCCESS in 46m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b74b1df55244a038fc7af05619ff7a1 : SUCCESS in 48m 43s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25e44648d00b4cf6a739a3c7edf1557d : SUCCESS in 39m 46s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/838ca118eb7a45cb884f45c06a738c66 : SUCCESS in 41m 04s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55f4922fbbd343f5a719df4f05691a62 : SUCCESS in 40m 15s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/19e78620d6654927a89bb09f33c8e399 : SUCCESS in 48m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dec72300c4fb4816920d7ff6b15eb881 : SUCCESS in 1h 06m 41s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8e02f2ac2194ae589964ea58216a5e3 : SUCCESS in 57m 26s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4acadf000777431096971a6ed38a132a : SUCCESS in 47m 10s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7fd28939804489cb05d0c0405b0a471 : SUCCESS in 48m 18s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/17377e6a8ef3433d92aa197dbe6b7662 : SUCCESS in 33m 54s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d075c654043748e0a0f28dae14c7726b : SUCCESS in 30m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4362fa34ffae4d69b7a42d691871dc2f : SUCCESS in 33m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/087d3323fc2e4a8796e320981689d4dc : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":3},{"id":"d6f72b2846ec801acf1ed181f96f18d2b7e31df9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-08 06:17:54.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4c7cb21b0bce16fe33873f7c8aeaa86acaa9bad2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 07:20:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d452f74a68c454b9e20565e8b1bf52a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b7662b6915348948842cdc3873d7451 : SUCCESS in 2m 23s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb7f3c0d268645a58d11fe0166e25449 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f54b4d12b41942379ea9984c4c9cf963 : SUCCESS in 5m 23s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/61855a20239b47de813b3b28664430e9 : SUCCESS in 11m 44s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d5c85b93e544140b8fca4d8af39ca08 : SUCCESS in 9m 26s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8bf6ce9310c43bba563843fd91eb769 : SUCCESS in 13m 03s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9de30f01d44e45cca9ed085f3c8fe45e : SUCCESS in 9m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6c70f23832741338959460d61e2ea92 : SUCCESS in 40m 18s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab9255b12db74af2b5153b456af88356 : SUCCESS in 56m 52s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52a4253954df4681aec74234ff1f2cc9 : SUCCESS in 43m 35s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eef2e5bda02a4ddc8b20904aa29c929a : SUCCESS in 35m 56s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81b5f39c8bdc4720831e040c440c0d85 : SUCCESS in 35m 44s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d12244229c2e4828a48fc68f334f6833 : SUCCESS in 42m 56s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3eda35f1dd6440f2b97067ff2b35743f : SUCCESS in 55m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b322cf8c847d47458fca7cb24f395384 : SUCCESS in 45m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7fa07e71a50946bea9f75fa5e16bd5e5 : SUCCESS in 49m 17s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2526071564e4f5d9420c1067e8826f7 : SUCCESS in 41m 15s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/08853be9c35e4126a22d4800c3c9dd3b : SUCCESS in 26m 18s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ba108d26eb344b38cffcf0d1362b13a : SUCCESS in 27m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af277bad3d0949b1b8c11d5e7d22a535 : SUCCESS in 31m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2e15bccc91e4b88b43f1b3a93bb7738 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":4},{"id":"f5065d07150d39757dbad10bfb733a42a149b31e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-05-08 11:55:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9332a6a9aa2528e8bf704a18d601698e5fc271ae","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-05-09 14:22:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6170f7f63dabd26ae2d1738dfc19f30d7d059e36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 14:23:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"eec292a82a47636244237bcc21274e53e4cdce4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 15:40:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb6b5279c19c46139c799cf1d39479df\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c142165acd664c1ebf672ca35a55b21d : SUCCESS in 3m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7b27f9dfd5a4c0c88eb68046c98f7b1 : SUCCESS in 2m 44s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/edf36a632a7440c194ff514c7fb9706f : SUCCESS in 10m 34s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b19b63633d75454c974b56ec725d7c2a : SUCCESS in 9m 50s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/de0d78d56c794d5db97a2dac322c9ad0 : SUCCESS in 54m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/08be89ea4f0e4266aa52de1f46326832 : SUCCESS in 1h 00m 46s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/64c6f37bb78c4f738c6fe37296b0b371 : SUCCESS in 53m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/52b1a9f71ad84b91aeda444dc81d1a8d : SUCCESS in 50m 28s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e70c7f51f3c64c5693fb26eacc9bbfa9 : SUCCESS in 50m 21s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/12bee93098644bc7bc19798a3c7b9dcf : SUCCESS in 35m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbc84f6604ea40e5a50c1da848ca26bd : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":4},{"id":"89a9e0aa1a9585c967f859390ece4c704cace5b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 15:40:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c095ac772d9789cd059619cd363b4104a7079a01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 15:41:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fb82eed18c9446a997e9bc743ae6dd1\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34dd1ab2cb1f4e37a8e601bcd14a075c : SUCCESS in 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d1b26af8e7b33c24ce31294dc0f060541f973b32","revisions":{"0abfb2c71e48d14f14f1eed7f3746030d7906c18":{"kind":"REWORK","_number":1,"created":"2024-05-08 03:23:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/918495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/95/918495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/1"}}},"commit":{"parents":[{"commit":"c5b4bcb1ae20e587c116bc1fa6f27b44c78ab8d8","subject":"Fix missing unit test coverage about [fernet_tokens] key_repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/c5b4bcb1ae20e587c116bc1fa6f27b44c78ab8d8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-08 03:22:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-08 03:22:43.000000000","tz":540},"subject":"Remove deprecated amqp messaging driver support","message":"Remove deprecated amqp messaging driver support\n\nIt was deprecated in 2024.1 release, so can be removed now.\n\nChange-Id: I0b5b54540242027707d94409be67fa1d73d037ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/0abfb2c71e48d14f14f1eed7f3746030d7906c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/0abfb2c71e48d14f14f1eed7f3746030d7906c18"}]},"branch":"refs/heads/master"},"6e8a3b91a645e96b8aa550b63737051ed9c7a383":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-05-08 03:28:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/918495/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/95/918495/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/2"}}},"commit":{"parents":[{"commit":"c5b4bcb1ae20e587c116bc1fa6f27b44c78ab8d8","subject":"Fix missing unit test coverage about [fernet_tokens] key_repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/c5b4bcb1ae20e587c116bc1fa6f27b44c78ab8d8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-08 03:22:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-08 03:28:00.000000000","tz":0},"subject":"Remove deprecated amqp messaging driver support","message":"Remove deprecated amqp messaging driver support\n\nIt was deprecated in 2023.2 release, so can be removed now.\n\nChange-Id: I0b5b54540242027707d94409be67fa1d73d037ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/6e8a3b91a645e96b8aa550b63737051ed9c7a383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/6e8a3b91a645e96b8aa550b63737051ed9c7a383"}]},"branch":"refs/heads/master","description":"Edit commit message"},"58f8fe2eeed6f9a0e3d8cc697ad3b01aa67c203d":{"kind":"REWORK","_number":3,"created":"2024-05-08 03:29:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/918495/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/95/918495/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/3"}}},"commit":{"parents":[{"commit":"c5b4bcb1ae20e587c116bc1fa6f27b44c78ab8d8","subject":"Fix missing unit test coverage about [fernet_tokens] key_repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/c5b4bcb1ae20e587c116bc1fa6f27b44c78ab8d8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-08 03:22:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-08 03:29:42.000000000","tz":540},"subject":"Remove deprecated amqp1 messaging driver support","message":"Remove deprecated amqp1 messaging driver support\n\nIt was deprecated in 2024.1 release, so can be removed now.\n\nChange-Id: I0b5b54540242027707d94409be67fa1d73d037ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/58f8fe2eeed6f9a0e3d8cc697ad3b01aa67c203d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/58f8fe2eeed6f9a0e3d8cc697ad3b01aa67c203d"}]},"branch":"refs/heads/master"},"d1b26af8e7b33c24ce31294dc0f060541f973b32":{"kind":"REWORK","_number":4,"created":"2024-05-08 06:17:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/918495/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/95/918495/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/95/918495/4"}}},"commit":{"parents":[{"commit":"c5b4bcb1ae20e587c116bc1fa6f27b44c78ab8d8","subject":"Fix missing unit test coverage about [fernet_tokens] key_repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/c5b4bcb1ae20e587c116bc1fa6f27b44c78ab8d8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-08 03:22:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-08 06:17:44.000000000","tz":540},"subject":"Remove deprecated amqp1 messaging driver support","message":"Remove deprecated amqp1 messaging driver support\n\nIt was deprecated in 2024.1 release, so can be removed now.\n\nChange-Id: I0b5b54540242027707d94409be67fa1d73d037ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d1b26af8e7b33c24ce31294dc0f060541f973b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d1b26af8e7b33c24ce31294dc0f060541f973b32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
