)]}'
{"id":"openstack%2Fpuppet-keystone~923194","triplet_id":"openstack%2Fpuppet-keystone~master~Ibc9becd17558cf66af4d9cf5ac8d7d8123daacc1","project":"openstack/puppet-keystone","branch":"master","topic":"oidc_additional_locations","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-17 11:54:23.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2024-07-22 15:00:28.000000000","reason":"removed on reply"},"36891":{"account":{"_account_id":36891,"name":"Francesco Di Nucci","display_name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","username":"d1nuc0m","status":"Sysasdmin @ INFN Italy"},"last_update":"2024-07-22 16:22:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibc9becd17558cf66af4d9cf5ac8d7d8123daacc1","subject":"feat: add param for additional OIDC locations","status":"MERGED","created":"2024-07-02 07:30:16.000000000","updated":"2024-07-22 16:23:46.000000000","submitted":"2024-07-22 16:22:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923194-oidc_additional_locations","meta_rev_id":"4f851176a2d99999eea54a9e4c56332ac4576733","_number":923194,"virtual_id_number":923194,"owner":{"_account_id":36891,"name":"Francesco Di Nucci","display_name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","username":"d1nuc0m","status":"Sysasdmin @ INFN Italy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-22 16:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-07-17 11:54:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-07-22 15:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-07-22 15:00:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 08:48:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-02 15:13:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-07-03 06:22:13.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"6536b2238432b3358720b6e17c43dd3858bfa41b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36891,"name":"Francesco Di Nucci","display_name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","username":"d1nuc0m","status":"Sysasdmin @ INFN Italy"},"date":"2024-07-02 07:30:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ece6beb56c8aa8a04fb92dcc9786a1eb304b4f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 08:48:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/deaecbfa162e44f5947943547cc37519\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7b2818c32b44c7dab72c56569acec1c : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad2b838623734cd09ab059a0a29dd409 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/00698128360845f8b64e0d33edddc0bf : SUCCESS in 4m 51s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/07cb319d795444638d83f90c870fc3a0 : FAILURE in 12m 46s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccecd097f8f148f9913a7ef18d078fec : FAILURE in 9m 01s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/a26049c62a9148f6bfcb8aa98551870e : FAILURE in 11m 43s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6036c032f8b54f7ca4762fecc30c593d : FAILURE in 9m 22s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3b9fc5c35314392b7eb87b5c0e4374e : SUCCESS in 40m 57s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5be68995335148dba5b0b0b39f9d83ff : SUCCESS in 1h 11m 27s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5315dbfc9d154c4fb9335f1941e44705 : SUCCESS in 37m 44s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6115ba5418646798182c9dc2f9bec88 : SUCCESS in 40m 08s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bd56b2d0716461d938fc03e44e0a7a3 : SUCCESS in 39m 59s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/01cc3ef6a276443c9c7e7bcb12986e5e : SUCCESS in 55m 03s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8222e4a14deb4b2e9bd14bbe6a6b729d : SUCCESS in 1h 08m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bcaa97ddb2154960b6a4a6e60a022eb3 : SUCCESS in 59m 07s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1512d879c76540d9959b17353ca65636 : SUCCESS in 50m 06s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd80241f05a245e692e61c32461ec651 : SUCCESS in 49m 16s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/0895055021d74073853cf5cf61163763 : SUCCESS in 33m 04s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba3bdfe78cc54376977f025c0f1732a6 : SUCCESS in 29m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5a6debf9c384b518edcb03e1993132f : SUCCESS in 28m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3e68ccf002348febf73b5c68d86899b : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4563efaac8e9e277ad048fbc300835a76696d881","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36891,"name":"Francesco Di Nucci","display_name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","username":"d1nuc0m","status":"Sysasdmin @ INFN Italy"},"date":"2024-07-02 09:03:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae5fefc35c4bee832f7cda33ab3b8c1b08827780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 11:10:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a00435e0a61743c2ba76d4a8961ca848\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e07c5a641d39455db51337859ac39ecd : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09c1e0380ed7457483e3b6b560945747 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/931073d99dbe4977b13a3c8626207479 : SUCCESS in 5m 45s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/b36959d47a684e28aa6d00e835f2fb4c : SUCCESS in 12m 45s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b4ca4ae7f954c76b7554eafa4ebc99b : SUCCESS in 10m 03s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b9241fdaab841269709519cfb547439 : SUCCESS in 13m 40s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f01f7d5c3094285b848d9e3f5178ab2 : SUCCESS in 8m 50s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aadd14c4163b49a393a9852c4b39935b : SUCCESS in 47m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1caa7ee49834347925d46e3a304ef60 : SUCCESS in 1h 04m 00s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b0634e127a449e4a912374c5573c385 : SUCCESS in 41m 56s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/528b99024e604491b82a901baabf3b81 : SUCCESS in 36m 32s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a736ea810ee4998813c6e7e88b2b5aa : SUCCESS in 43m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b9a5b7b7075495489d42decb9ab6248 : SUCCESS in 54m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/32e8ed9cf971410fa25e4df0bde7dbce : SUCCESS in 1h 26m 10s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf6f5cc3a3694b4fa13b5210e6acab46 : SUCCESS in 56m 26s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/de1d316349664de5a7780d664788a7fb : SUCCESS in 1h 02m 43s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fbbb8f0ade94adeb2ce9f332e7ef06b : SUCCESS in 50m 46s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/3014ce43dd034917aecd2e0ddaea464f : SUCCESS in 32m 32s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06a0707b35ac46edb808b1788b2fe80d : SUCCESS in 32m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be03030f0e15484eb170df25eca12f23 : SUCCESS in 32m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3c6f40308534687afcdd6e075deabcb : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":2},{"id":"3968d572ba14047aacc3e17c76e3454a50574c6e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-07-02 15:13:13.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0ea2ca984130372db3d7f467da97b72a486431ae","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-03 06:22:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8fd71dd0ba7d5db7898c4aa6c2a62fda2cd354e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36891,"name":"Francesco Di Nucci","display_name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","username":"d1nuc0m","status":"Sysasdmin @ INFN Italy"},"date":"2024-07-04 10:22:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ee34cc37e8965f011ec056466603c570d6c29ef7","author":{"_account_id":36891,"name":"Francesco Di Nucci","display_name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","username":"d1nuc0m","status":"Sysasdmin @ INFN Italy"},"date":"2024-07-04 10:22:20.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d3b5f7b03aff19bb5421756cbb43fc918a61bfd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 12:48:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/994b8ca496774f5d99bde5224d05fbd8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77b5b62b23e9485c885643fd9c03d943 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4299b3d79c3745beb1b520b90c76d985 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/092ac858fa664eaf9c4c98f1586b5d2a : SUCCESS in 5m 30s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d5668872ca645d2bb4ff8e9db0dbbd5 : SUCCESS in 13m 36s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9eae70fea044180a1ea25b93451c055 : SUCCESS in 10m 05s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/8dab02924b774e51b0d2f55f9cb14c92 : SUCCESS in 12m 49s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bd5eb5d3a6945d6b2718fd3f88516cb : SUCCESS in 9m 59s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67eceb700c8b47e38097ca04c9915230 : SUCCESS in 46m 56s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebf9933e20a7468cb9c987e3cbc5f74e : FAILURE in 55m 15s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e93deafa72854574a06f116391ee52a4 : SUCCESS in 44m 19s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d03f5ae6ea444b5eb33e7e0e990461c4 : FAILURE in 51m 06s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00e33badcff244398cca15b626316637 : SUCCESS in 41m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8aed78556d74417abf63537100c196ae : SUCCESS in 1h 05m 19s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/477c22fa3534464c83ebe477f0b9300a : SUCCESS in 1h 31m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c02c99e9840b4184b720523812460a1c : SUCCESS in 1h 02m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/890a62e057d6451994e49b1395dc9cda : SUCCESS in 1h 10m 44s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/29541e2847924c73ac3897f319d9f252 : SUCCESS in 1h 05m 27s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a37e3fc25314257b6b8bf1ff3a480d0 : SUCCESS in 39m 14s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48cad4d1bcb340c88578f63ddf4191e0 : SUCCESS in 37m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02339dd1ee4248c983246f805258701c : SUCCESS in 34m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f81a9235ffe64d189dac3cf8e84b662c : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":3},{"id":"f5c7c2c45cca96b373979c964affa1ec2a77534a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-04 13:38:24.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"4ba784d346ec3c2e4bcbb659efae197941c3dd99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 15:10:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a61b55bda9c94069a0c944b212b106fe\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fbc79334f584a7dbd277ea230494778 : SUCCESS in 2m 30s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab3289b0e2da4abfa43b970bbb3651df : SUCCESS in 3m 02s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/39284a878dd84df09298b5ed9df9f2b0 : SUCCESS in 6m 51s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/5dcb8f559eb24e01a8e08624b14bc230 : SUCCESS in 13m 41s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ea06744ad414e71b657ddc5dbef9818 : SUCCESS in 8m 47s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/2385d6b7887e4c4395dcc9e3479b1d72 : SUCCESS in 13m 28s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b595162d8c594ddf8a4cdc1dc2a0f26f : SUCCESS in 10m 14s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23bcc29b78754bf3aeac53641c147755 : SUCCESS in 49m 35s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fcd3d4d5ed2d48abb64f830dd13cc818 : FAILURE in 54m 39s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7514c44d63f14fb8a7553ed822af7aa0 : SUCCESS in 54m 04s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec8f67a3bc00494a916af0d062e3030e : SUCCESS in 46m 59s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c1380188dd64807857111ca9014ac76 : SUCCESS in 39m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d504345f040405aa8aa56948024ff07 : SUCCESS in 59m 01s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d728f0d2e5544a2a2a50ce57ae063ee : SUCCESS in 1h 14m 39s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/25a0b8cde01d45a3bfedd6c83ded14cd : SUCCESS in 1h 00m 46s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d3bc4e8475841fba41d48463409633e : SUCCESS in 1h 02m 45s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/23d6d272be4741649c02eafb133efdcd : SUCCESS in 1h 00m 59s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac04aa7e870d4e849776bcb0aabbc35d : SUCCESS in 35m 40s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c52003c4531946abbdfdd9861fed6951 : SUCCESS in 34m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f8a223b8b154a40891f127aabb2fb6a : SUCCESS in 33m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86fe4f2cbdfe404c9016679e303d74b0 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":4},{"id":"8600b168d2a9e0e520d294453516fe13fc1d8999","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-07-10 21:34:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b638b32b6cd6f65ff9e63fc2bb9f0b6fb30a2ec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36891,"name":"Francesco Di Nucci","display_name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","username":"d1nuc0m","status":"Sysasdmin @ INFN Italy"},"date":"2024-07-11 08:30:18.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"eb0378ead73d5a754796e9d25a2efd7fd133c1f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 09:55:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4498ef3c8e0e40abb33eec9bd3cc7318\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50af8ae0c8104b7a90d2706d8f7b8d0c : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/648912f86d964788b93153e289c2879f : SUCCESS in 2m 47s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a1be1e797ff44958c547e3053075968 : SUCCESS in 4m 39s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/201d58f82ed84bf5bf157dcd6f48e080 : SUCCESS in 11m 46s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5cb9b845c9de484db9626083eb888c88 : SUCCESS in 10m 48s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/a908adc4b0b543f884377d4857d0d30d : SUCCESS in 11m 34s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c902526ce5e24dcb9ff86c9d49183afd : SUCCESS in 8m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b9195dc1ea64041946a7969ab10bbed : SUCCESS in 49m 26s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45a95d67de4d49de9ce7896412fd55e5 : SUCCESS in 50m 07s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd97b12b7dfe47b7b159dda28912caac : SUCCESS in 46m 53s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc92d430415842a6a535d4141c22f6bf : SUCCESS in 42m 17s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11e9805fcad54e089cfd1a774125b9bf : SUCCESS in 42m 58s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7acac6bd3cc449e94262e36d84bdffc : SUCCESS in 51m 35s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9810a67e77d944619e792b4649a10fb6 : SUCCESS in 1h 19m 01s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/124089c6378242a5a1e702c25e87c93d : SUCCESS in 53m 16s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f001f27335f04121abd68bffee5bde1e : SUCCESS in 1h 01m 59s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1265d988587847059c9fba1cacd50416 : SUCCESS in 53m 47s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/e29280c5a53d45ffa53ef43eebb7193f : SUCCESS in 28m 10s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ff1421ba0b84ddb96063f7a091449a2 : SUCCESS in 30m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62bf2e406edb446a9ae02d581a07c899 : SUCCESS in 32m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/082e582ddfd54e0abd8043b8627e7391 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":5},{"id":"0807956c331aad797bf92156ccbcb7632a8619ce","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-11 14:28:29.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b5667993aca83a283257e64d1caec1b12cf219c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36891,"name":"Francesco Di Nucci","display_name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","username":"d1nuc0m","status":"Sysasdmin @ INFN Italy"},"date":"2024-07-17 08:30:15.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"b0e89278d8791737de9de1474742643b74169e2d","author":{"_account_id":36891,"name":"Francesco Di Nucci","display_name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","username":"d1nuc0m","status":"Sysasdmin @ INFN Italy"},"date":"2024-07-17 08:30:56.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"608e3ebb0f4bc01a785b620e02322aecafaa566f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 09:50:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/941e780b522844ba8d2bb515b5a68c4c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43bc17ee2c4f4c8ca848a5ec0e6fc487 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aed91ae77a764ee793472eea8651048c : SUCCESS in 3m 35s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f18695c8820b49269bcc524815d294aa : SUCCESS in 4m 54s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/37f9d03f83544ecfbde6c320fc905528 : SUCCESS in 11m 17s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc15354a87934551b1f098cbbf75251b : SUCCESS in 9m 40s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a617770d0a4461ea7c52f3f5a11c330 : SUCCESS in 12m 01s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dac9bc8cd9674f3291d521e2b5d38da2 : SUCCESS in 9m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed18f3f938544f5b9d108304e5a8ddbd : SUCCESS in 38m 53s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c961066b0bb149e08a82d323192eb381 : SUCCESS in 1h 12m 24s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7cfeca0f39243488fbb7d0c4a4d9161 : SUCCESS in 49m 49s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02b02a30b76c4568b6737775773694c9 : SUCCESS in 43m 32s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc0ec68c659b440c9a00476c57a1d442 : SUCCESS in 42m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/656de1803a4d465195320ad138d704ac : SUCCESS in 56m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/25eb6d3a875b4de29ff9f02dd313c555 : SUCCESS in 1h 18m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/386d537aec0042a7858bf118a77c6e1f : SUCCESS in 57m 56s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bd2bafb1048442791c3d68a9d3da74b : SUCCESS in 56m 47s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d92322a107b4053bceb0abfb5342dcf : SUCCESS in 55m 23s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/c61383f515234156a1ffebaa3f8b8a59 : SUCCESS in 40m 27s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf9b28ec967e40c99a94ccbe2cbbdc64 : SUCCESS in 30m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abefc12624c54ceaab02140c8367d770 : SUCCESS in 30m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b91b7b98b9c4d48863ec7816e05ea5b : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":6},{"id":"4a518eb77f9649770be34d7be3e1a54684afc1ae","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-17 11:54:23.000000000","message":"Patch Set 6: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"33297476721487c1142746e3d84eb7435d301a59","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-07-22 15:00:28.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"a5030b5b8bc0475639e77173326302ea59898ddc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 15:01:20.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ab2b9aaba57ff84b8462897719fd69c7a6226656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:22:43.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3439ec248e9a4fed942f5139e7585b6c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/060ea7b399344ad1a46154a125845437 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5bd71c370ff48f1a4e96998cee193d4 : SUCCESS in 4m 57s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/32d9fb0dcb3b40119e5582183c308f77 : SUCCESS in 11m 39s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84e3811be5024bfe8970e3e6864f4c28 : SUCCESS in 10m 00s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7621c8e068de4b5f8ce1fb969ee4c6ec : SUCCESS in 56m 46s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b981b666270b4587ac04dad4542a08ad : SUCCESS in 1h 20m 35s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d507aa88f2384d379ae36e924cb0d192 : SUCCESS in 56m 46s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d52eddbf38b3446e82c7cd5d409568ed : SUCCESS in 1h 03m 14s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/de3d8a3bb8e0407a890edc164297543c : SUCCESS in 56m 30s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4b8664e031c42c5bcb0a5bc933f0e99 : SUCCESS in 39m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/114577b9d6504e5e949013adb4d0a349 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":6},{"id":"5dacf3899df937f4d2e1262692ae05ff08bbd3d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:22:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"4f851176a2d99999eea54a9e4c56332ac4576733","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:23:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6a5aa1976014e3191ee209344b2488a\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/783ec632e2a64ba28c2c6dc49ceab861 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"81d2a80f0a75dccb0c7aabdef7abd91b1f375f20","revisions":{"8e21dfe30d7255a5224a36da26217fbe1014c7e9":{"kind":"REWORK","_number":1,"created":"2024-07-02 07:30:16.000000000","uploader":{"_account_id":36891,"name":"Francesco Di Nucci","display_name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","username":"d1nuc0m","status":"Sysasdmin @ INFN Italy"},"ref":"refs/changes/94/923194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/94/923194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/1"}}},"commit":{"parents":[{"commit":"68e091978887937a46bf9dd389e4b9d85079ac59","subject":"feat: add a param for arbitrary federation OIDCRedirectURI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/68e091978887937a46bf9dd389e4b9d85079ac59"}]}],"author":{"name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","date":"2024-07-02 06:08:01.000000000","tz":120},"committer":{"name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","date":"2024-07-02 07:26:24.000000000","tz":120},"subject":"feat: add param for additional OIDC locations","message":"feat: add param for additional OIDC locations\n\nAdd keystone::federation::openidc:locations to add extra Location\ndirectives in OIDC WSGI config\n\nChange-Id: Ibc9becd17558cf66af4d9cf5ac8d7d8123daacc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/8e21dfe30d7255a5224a36da26217fbe1014c7e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/8e21dfe30d7255a5224a36da26217fbe1014c7e9"}]},"branch":"refs/heads/master"},"50398f815d83486c41d585a602b39e9026c17f39":{"kind":"REWORK","_number":2,"created":"2024-07-02 09:03:14.000000000","uploader":{"_account_id":36891,"name":"Francesco Di Nucci","display_name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","username":"d1nuc0m","status":"Sysasdmin @ INFN Italy"},"ref":"refs/changes/94/923194/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/94/923194/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/2"}}},"commit":{"parents":[{"commit":"68e091978887937a46bf9dd389e4b9d85079ac59","subject":"feat: add a param for arbitrary federation OIDCRedirectURI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/68e091978887937a46bf9dd389e4b9d85079ac59"}]}],"author":{"name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","date":"2024-07-02 06:08:01.000000000","tz":120},"committer":{"name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","date":"2024-07-02 09:02:43.000000000","tz":120},"subject":"feat: add param for additional OIDC locations","message":"feat: add param for additional OIDC locations\n\nAdd keystone::federation::openidc:locations to add extra Location\ndirectives in OIDC WSGI config\n\nChange-Id: Ibc9becd17558cf66af4d9cf5ac8d7d8123daacc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/50398f815d83486c41d585a602b39e9026c17f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/50398f815d83486c41d585a602b39e9026c17f39"}]},"branch":"refs/heads/master"},"4226839c961ab6fd0316a4bea1afc0b13be35897":{"kind":"REWORK","_number":3,"created":"2024-07-04 10:22:08.000000000","uploader":{"_account_id":36891,"name":"Francesco Di Nucci","display_name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","username":"d1nuc0m","status":"Sysasdmin @ INFN Italy"},"ref":"refs/changes/94/923194/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/94/923194/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/3"}}},"commit":{"parents":[{"commit":"68e091978887937a46bf9dd389e4b9d85079ac59","subject":"feat: add a param for arbitrary federation OIDCRedirectURI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/68e091978887937a46bf9dd389e4b9d85079ac59"}]}],"author":{"name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","date":"2024-07-02 06:08:01.000000000","tz":120},"committer":{"name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","date":"2024-07-04 10:21:59.000000000","tz":120},"subject":"feat: add param for additional OIDC locations","message":"feat: add param for additional OIDC locations\n\nAdd keystone::federation::openidc:locations to add extra Location\ndirectives in OIDC WSGI config\n\nChange-Id: Ibc9becd17558cf66af4d9cf5ac8d7d8123daacc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/4226839c961ab6fd0316a4bea1afc0b13be35897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/4226839c961ab6fd0316a4bea1afc0b13be35897"}]},"branch":"refs/heads/master"},"3b3587096d9d785c20aaf139c529ed37d5f28b4c":{"kind":"REWORK","_number":4,"created":"2024-07-04 13:38:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/923194/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/94/923194/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/4"}}},"commit":{"parents":[{"commit":"68e091978887937a46bf9dd389e4b9d85079ac59","subject":"feat: add a param for arbitrary federation OIDCRedirectURI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/68e091978887937a46bf9dd389e4b9d85079ac59"}]}],"author":{"name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","date":"2024-07-02 06:08:01.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-04 13:38:07.000000000","tz":540},"subject":"feat: add param for additional OIDC locations","message":"feat: add param for additional OIDC locations\n\nAdd keystone::federation::openidc::additional_locations to add extra\nLocation directives in OIDC WSGI config\n\nChange-Id: Ibc9becd17558cf66af4d9cf5ac8d7d8123daacc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/3b3587096d9d785c20aaf139c529ed37d5f28b4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/3b3587096d9d785c20aaf139c529ed37d5f28b4c"}]},"branch":"refs/heads/master"},"935d507867b436e8354e29e1692dfbb9ed38e590":{"kind":"REWORK","_number":5,"created":"2024-07-11 08:30:18.000000000","uploader":{"_account_id":36891,"name":"Francesco Di Nucci","display_name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","username":"d1nuc0m","status":"Sysasdmin @ INFN Italy"},"ref":"refs/changes/94/923194/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/94/923194/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/5"}}},"commit":{"parents":[{"commit":"e4acfc9f29e17d17ad736d733c4989b31a4d7799","subject":"feat: add openidc_metadata_dir to federation::oidc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/e4acfc9f29e17d17ad736d733c4989b31a4d7799"}]}],"author":{"name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","date":"2024-07-02 06:08:01.000000000","tz":120},"committer":{"name":"Francesco Di Nucci","email":"francesco.dinucci@na.infn.it","date":"2024-07-11 08:29:47.000000000","tz":120},"subject":"feat: add param for additional OIDC locations","message":"feat: add param for additional OIDC locations\n\nAdd keystone::federation::openidc::additional_locations to add extra\nLocation directives in OIDC WSGI config\n\nChange-Id: Ibc9becd17558cf66af4d9cf5ac8d7d8123daacc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/935d507867b436e8354e29e1692dfbb9ed38e590"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/935d507867b436e8354e29e1692dfbb9ed38e590"}]},"branch":"refs/heads/master"},"81d2a80f0a75dccb0c7aabdef7abd91b1f375f20":{"kind":"REWORK","_number":6,"created":"2024-07-17 08:30:15.000000000","uploader":{"_account_id":36891,"name":"Francesco Di Nucci","display_name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","username":"d1nuc0m","status":"Sysasdmin @ INFN Italy"},"ref":"refs/changes/94/923194/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/94/923194/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/94/923194/6"}}},"commit":{"parents":[{"commit":"e4acfc9f29e17d17ad736d733c4989b31a4d7799","subject":"feat: add openidc_metadata_dir to federation::oidc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/e4acfc9f29e17d17ad736d733c4989b31a4d7799"}]}],"author":{"name":"Francesco Di Nucci","email":"d1nuc0m@protonmail.com","date":"2024-07-02 06:08:01.000000000","tz":120},"committer":{"name":"Francesco Di Nucci","email":"francesco.dinucci@na.infn.it","date":"2024-07-17 08:29:46.000000000","tz":120},"subject":"feat: add param for additional OIDC locations","message":"feat: add param for additional OIDC locations\n\nAdd keystone::federation::openidc::additional_locations to add extra\nLocation directives in OIDC WSGI config\n\nChange-Id: Ibc9becd17558cf66af4d9cf5ac8d7d8123daacc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/81d2a80f0a75dccb0c7aabdef7abd91b1f375f20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/81d2a80f0a75dccb0c7aabdef7abd91b1f375f20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
