)]}'
{"id":"openstack%2Fpuppet-keystone~923703","triplet_id":"openstack%2Fpuppet-keystone~stable%2F2024.1~I2c5410408dd3fd837e55a538e46de4f4fe5127be","project":"openstack/puppet-keystone","branch":"stable/2024.1","hashtags":[],"change_id":"I2c5410408dd3fd837e55a538e46de4f4fe5127be","subject":"Fix invalid string concatenation","status":"MERGED","created":"2024-07-09 04:56:00.000000000","updated":"2024-07-09 08:10:33.000000000","submitted":"2024-07-09 08:10:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923703","cherry_pick_of_change":923238,"cherry_pick_of_patch_set":2,"meta_rev_id":"2b7c91c053b4a9596624064f4c87195eb7da350c","_number":923703,"virtual_id_number":923703,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-09 08:10:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-07-09 06:33:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-07-09 06:33:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-09 06:21:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ee58d0ed6c3c10aefad1e258c303ce73571f80d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-09 04:56:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"208a155fc3fa0b861f5cfd5b68ed788f8e39edec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 06:21:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0dc7e1696f84af2b33f7dc11cd230f0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ee54eba09014dc3872f7d813c14a722 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dde099aafee42d98a8f55b8ecb2b3a6 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f9a8dd7359c45dabce293855e2a014e : SUCCESS in 4m 40s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8b68677e8b94b0494d70b87d5eda006 : SUCCESS in 10m 16s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91be856a50914d5da2c9f1a545f43753 : SUCCESS in 9m 23s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/332b0fabb2da4a59b927ca9fd0155444 : SUCCESS in 13m 05s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c91d88a40ba741d98d8f9d0fb8fc03d7 : SUCCESS in 9m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62fa18de99554b3cb0bc3b7bfcd9f359 : SUCCESS in 39m 14s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e03977b82be459092377f6f677f3334 : SUCCESS in 1h 00m 42s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/322add2a8e46443585f8291fc241bdac : SUCCESS in 45m 14s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fccc551cad0344d4b23ac1c47f1a13b0 : SUCCESS in 38m 26s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1ba35b5c61540acbca3247bc2d31e33 : FAILURE in 46m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af5c248bfe024ced965497533e81fb83 : SUCCESS in 55m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f873d45b331c414cb3ba65b20d8144af : SUCCESS in 1h 19m 48s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/79062a1323534e35b0a221402414d12f : SUCCESS in 58m 36s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba0b09783d154328b058847897dc5d6d : SUCCESS in 48m 15s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d469cd5a84974ef2b823a7452a604e9e : SUCCESS in 51m 13s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/226efc61e2ef4497922752701284b49a : SUCCESS in 28m 44s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35ca8b9e8d90417e96d547929ba07145 : SUCCESS in 32m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6a66b3b59d24cbf835a216d06210b60 : SUCCESS in 32m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"134199492fb67dbe1bdcac2f1856fc378ff7ffed","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-09 06:33:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f46318e63a67aef66472a45dbbc961d4ddd75863","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 06:34:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"685d83ee2862d93cd71eec805fdde5d77ebabd78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 08:10:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/193da3fdfae34726982c25a90e28e3f2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4dc53820a3f542d985ab12888148f6f9 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6520816434944aa9b6c8c8e4235c2888 : SUCCESS in 2m 33s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/799d37dde75a49a39a8c3b0d7a80e3e5 : SUCCESS in 13m 22s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad3229bd11e64195848b255ca80c11d3 : SUCCESS in 11m 06s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/536d7a5c7afa440baa77f3e5b257d2b8 : SUCCESS in 1h 00m 18s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/487b6bac3e084fae9dac7b058f485d5d : SUCCESS in 1h 22m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1073850b094343d5bafd5b84c57d0570 : SUCCESS in 1h 06m 13s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb0d361f11fc40eaab80f547e1007f7d : SUCCESS in 1h 01m 47s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/17a500d94d264231bcc02063068ab4cd : SUCCESS in 1h 02m 52s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/717980d8b4f34315a302950d8f975cdf : SUCCESS in 39m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2b7c91c053b4a9596624064f4c87195eb7da350c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 08:10:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"753b0f8e0599ddf71bfbbcf40449b6f9192d5ff6","revisions":{"753b0f8e0599ddf71bfbbcf40449b6f9192d5ff6":{"kind":"REWORK","_number":1,"created":"2024-07-09 04:56:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/923703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/03/923703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/03/923703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/03/923703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/03/923703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/03/923703/1"}}},"commit":{"parents":[{"commit":"23db16b256319c3c0491c7fb2710d1100cb20d18","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/23db16b256319c3c0491c7fb2710d1100cb20d18"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-02 13:22:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-09 04:56:00.000000000","tz":0},"subject":"Fix invalid string concatenation","message":"Fix invalid string concatenation\n\nPuppet does not support concatenating strings by + .\n\nChange-Id: I2c5410408dd3fd837e55a538e46de4f4fe5127be\n(cherry picked from commit 7558e5bfa19399e969401a7fe97feea5cce4430f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/753b0f8e0599ddf71bfbbcf40449b6f9192d5ff6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/753b0f8e0599ddf71bfbbcf40449b6f9192d5ff6"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
