)]}'
{"id":"openstack%2Fpuppet-keystone~938359","triplet_id":"openstack%2Fpuppet-keystone~master~I9be33b79d94d8a7c85b8e1109659e38c8fb6245e","project":"openstack/puppet-keystone","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-02 13:34:54.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9be33b79d94d8a7c85b8e1109659e38c8fb6245e","subject":"Revert \"Disable a few acceptance test scenarios\"","status":"NEW","created":"2025-01-02 13:09:34.000000000","updated":"2025-02-22 16:38:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b97253d793cf7990f8c544031368d2df6e1122e4","_number":938359,"virtual_id_number":938359,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-22 16:38:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-02 13:34:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"330f5bba2e0d4ffa8ea735bb26a4db5a5939219f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-02 13:09:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c1a9216e648e3f19d8832c9e6493a979fc1a9d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 13:34:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9a75309c0d64c41b2a5c7c5f9ec353a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ba8da6033fa46b28047897394f97466 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bbcbd94d83d4cd1bcd59b58bbe99205 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0701baec2644e4c9fb774c3a3e031ed : RETRY_LIMIT in 1m 35s (non-voting)\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/397301a87edb4c3cb208bc052ab8725d : RETRY_LIMIT in 2m 10s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfe7a08b78374606af58afe194cfa745 : SUCCESS in 22m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7731ea7b5be1485f9dcbeea9caf63b7b : SUCCESS in 24m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b07c7da4db70a5e5787a915aa9d9f70b379b477c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-09 13:28:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d059e240b2bcb7a4876a3e13893675cae122b8b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-09 13:40:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"151e674fcd1920fcd0a986b2d20f9b97b7662354","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-09 15:15:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f922df37da6d41c88dcf1c38edbf00c3\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e1cbe308ff647e28c0045531390f61e : SUCCESS in 4m 48s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a44387caf7984f49ad6ac83e4c608431 : SUCCESS in 5m 23s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e62b82e65154b0c9e8655ddf7dc891c : SUCCESS in 3m 25s (non-voting)\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/23eedf281b764ab68a26a7179249463c : TIMED_OUT in 1h 33m 33s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c99f3a5f6e4b4e3b81ea04252f8d5e03 : SUCCESS in 33m 09s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/a066f0724e8e416c9f1f72494088ff6d : TIMED_OUT in 1h 31m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dd2cd9e79e24472ae275f7e4ff29c5d : SUCCESS in 30m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3e16fcad87dcff5e69281aebdb3ce9389eade32f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-19 15:29:08.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"3cea89d01b9b55f57eaffff221e5609018cbc535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 17:09:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ee48111a0d94ebba9d14533f04e2629\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f8551c57ee746b1925ad10a4ad7ef40 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2070695ad24c4ab99cf0c9cddf3a77c9 : SUCCESS in 2m 53s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a0575162bc149b6befd700df878e6e2 : SUCCESS in 6m 15s (non-voting)\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf6f03802fbb4d7b83034140a9320dc8 : TIMED_OUT in 1h 33m 53s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/533d3a6712d143318c45fe7fba97fd72 : SUCCESS in 34m 00s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/99edf48f1e3b4ada8daa2e8f94ccad78 : TIMED_OUT in 1h 33m 45s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/230626c4d381487399183b6ae8ff4736 : SUCCESS in 30m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5e42dc477d6a4ca0c45e9cdef11f202461882a78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-22 15:03:33.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b97253d793cf7990f8c544031368d2df6e1122e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 16:38:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8932a34dc2e74055ace741b5e05ffca9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9730242cdfd470588a9e9b65a484029 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7b849175b524c19be0cf645a518974c : SUCCESS in 2m 49s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/94902248439349eeb9c1f5e7e21800a4 : SUCCESS in 5m 46s (non-voting)\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/9effae63667948fc84c3476538875987 : TIMED_OUT in 1h 33m 06s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1ca9efbc61540658611ff206f5c3b9b : SUCCESS in 16m 27s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5105c17865948cfa1dfde0eafade719 : TIMED_OUT in 1h 33m 18s\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12665fb99d574e09a4653f75ce98b977 : SUCCESS in 26m 34s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"68d62b70be3facb159e0cd754f9db4774ff3b40f","revisions":{"519d4751fea32fe4a9b1c7c207f08efd95f47949":{"kind":"REWORK","_number":1,"created":"2025-01-02 13:09:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/938359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/59/938359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/1"}}},"commit":{"parents":[{"commit":"327c0a53d549b9b7cf69b552f6549c698cd15342","subject":"Merge \"Add support for [cache] backend_expiration_time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/327c0a53d549b9b7cf69b552f6549c698cd15342"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-02 13:09:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-02 13:09:19.000000000","tz":540},"subject":"Revert \"Disable a few acceptance test scenarios\"","message":"Revert \"Disable a few acceptance test scenarios\"\n\nThis reverts commit eedb1057206684507810baa79a672e9d78e8aeb4.\n\nReason for revert:\nNew python-openstackclient release fixed the issue.\n\nChange-Id: I9be33b79d94d8a7c85b8e1109659e38c8fb6245e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/519d4751fea32fe4a9b1c7c207f08efd95f47949"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/519d4751fea32fe4a9b1c7c207f08efd95f47949"}]},"branch":"refs/heads/master"},"90f384bf9b00869ebd18b968de3ccc2bb8ece11a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-09 13:28:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/938359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/59/938359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/2"}}},"commit":{"parents":[{"commit":"319b346e2da66d28ad743969279ccdec646d08e9","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/319b346e2da66d28ad743969279ccdec646d08e9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-02 13:09:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-09 13:28:28.000000000","tz":0},"subject":"Revert \"Disable a few acceptance test scenarios\"","message":"Revert \"Disable a few acceptance test scenarios\"\n\nThis reverts commit eedb1057206684507810baa79a672e9d78e8aeb4.\n\nReason for revert:\nNew python-openstackclient release fixed the issue.\n\nChange-Id: I9be33b79d94d8a7c85b8e1109659e38c8fb6245e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/90f384bf9b00869ebd18b968de3ccc2bb8ece11a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/90f384bf9b00869ebd18b968de3ccc2bb8ece11a"}]},"branch":"refs/heads/master","description":"Rebase"},"2760a1d1a79aac0b280bcbbdb3bf9d3e712a5d0b":{"kind":"REWORK","_number":3,"created":"2025-02-09 13:40:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/938359/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/59/938359/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/3"}}},"commit":{"parents":[{"commit":"319b346e2da66d28ad743969279ccdec646d08e9","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/319b346e2da66d28ad743969279ccdec646d08e9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-02 13:09:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-09 13:40:27.000000000","tz":540},"subject":"Revert \"Disable a few acceptance test scenarios\"","message":"Revert \"Disable a few acceptance test scenarios\"\n\nThis reverts commit eedb1057206684507810baa79a672e9d78e8aeb4.\n\nReason for revert:\nNew python-openstackclient release fixed the issue.\n\nChange-Id: I9be33b79d94d8a7c85b8e1109659e38c8fb6245e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/2760a1d1a79aac0b280bcbbdb3bf9d3e712a5d0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/2760a1d1a79aac0b280bcbbdb3bf9d3e712a5d0b"}]},"branch":"refs/heads/master"},"b939a48196a8548b61346fe68a6dbbb0e438804f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-02-19 15:29:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/938359/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/59/938359/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/4"}}},"commit":{"parents":[{"commit":"18df31d8faccff8d7d3d7d051a7825a18d7dec9e","subject":"Merge \"Mark puppet 8 supported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/18df31d8faccff8d7d3d7d051a7825a18d7dec9e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-02 13:09:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-19 15:29:08.000000000","tz":0},"subject":"Revert \"Disable a few acceptance test scenarios\"","message":"Revert \"Disable a few acceptance test scenarios\"\n\nThis reverts commit eedb1057206684507810baa79a672e9d78e8aeb4.\n\nReason for revert:\nNew python-openstackclient release fixed the issue.\n\nChange-Id: I9be33b79d94d8a7c85b8e1109659e38c8fb6245e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/b939a48196a8548b61346fe68a6dbbb0e438804f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/b939a48196a8548b61346fe68a6dbbb0e438804f"}]},"branch":"refs/heads/master","description":"Rebase"},"68d62b70be3facb159e0cd754f9db4774ff3b40f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-02-22 15:03:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/938359/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/59/938359/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/59/938359/5"}}},"commit":{"parents":[{"commit":"d4ada5b65466d60632232c34771c5d91d6687c97","subject":"feat: add openidc_scope param to set OIDCScope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d4ada5b65466d60632232c34771c5d91d6687c97"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-02 13:09:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-02-22 15:03:33.000000000","tz":0},"subject":"Revert \"Disable a few acceptance test scenarios\"","message":"Revert \"Disable a few acceptance test scenarios\"\n\nThis reverts commit eedb1057206684507810baa79a672e9d78e8aeb4.\n\nReason for revert:\nNew python-openstackclient release fixed the issue.\n\nChange-Id: I9be33b79d94d8a7c85b8e1109659e38c8fb6245e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/68d62b70be3facb159e0cd754f9db4774ff3b40f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/68d62b70be3facb159e0cd754f9db4774ff3b40f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
