)]}'
{"id":"openstack%2Fpuppet-keystone~947454","triplet_id":"openstack%2Fpuppet-keystone~master~Iff16deb64b2b2776aa5deab4d27247b8d58022ac","project":"openstack/puppet-keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-04-17 09:15:12.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2025-04-16 21:46:40.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Iff16deb64b2b2776aa5deab4d27247b8d58022ac","subject":"Remove watch_log_file","status":"MERGED","created":"2025-04-16 11:54:27.000000000","updated":"2025-04-17 09:15:59.000000000","submitted":"2025-04-17 09:15:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947454","meta_rev_id":"0a101b3cbd0362a482a540353569e9595a46667d","_number":947454,"virtual_id_number":947454,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-17 09:15:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-04-16 13:04:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-04-16 21:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-16 21:02:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-04-16 21:46:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-16 13:02:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-16 13:04:49.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-04-16 21:02:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-04-16 21:03:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"853f445e4680b9169d4f44ecce9bb1a7e5d6e319","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-16 11:54:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"752768e8718f99a437e190b74b4cf6ed4477bdd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 13:02:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea310b0d289547eb82cb710689444df7\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c58e4b93cd1f411cafc0c93c2e1267d8 : SUCCESS in 2m 16s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b0d063e47a8450fb0cca0a5427dc0eb : SUCCESS in 2m 37s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73b5229ec5dc459aaaa839351f7212fc : SUCCESS in 1m 33s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/97e07b238cdc42f8b73fd8ccb9e7124b : SUCCESS in 11m 18s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c889cac46a544d7ab1671814f0b0eb6a : SUCCESS in 8m 03s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25615ab45cc941d9a531232b22bf9b7a : SUCCESS in 7m 48s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85ee4100dc074870ac2946a72f02ca88 : SUCCESS in 50m 18s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdb38c96b10640d1810449743a7db25b : SUCCESS in 58m 55s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb8e310999144adcaa65b42018183978 : SUCCESS in 41m 52s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5b028ba11eb4fe38becf2eb33e5235a : SUCCESS in 54m 41s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00874595833c4a45a89611ce9ecd32f9 : FAILURE in 26m 55s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee4fefb254d94771ab8140419e0a1931 : SUCCESS in 1h 03m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/18e5f3dee6194970a34406689a7a7fb7 : SUCCESS in 45m 32s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2537dd6b84845c780ddb79bfe9e0abb : SUCCESS in 52m 20s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/18c24bbb6fac490b962b76c3038bca1c : SUCCESS in 52m 54s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/39b6ce2a056644fa92138219b237042f : SUCCESS in 55m 11s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/83d6b6d911b14d89b61d68f37e084686 : SUCCESS in 22m 21s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95d231720e4c4673a604bbdeaf59ec45 : SUCCESS in 20m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be6d93364d16456e8c31e317ae0f5c39 : SUCCESS in 19m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38921198a70c40de8b0e23bd37ae9911 : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ccbf48634a716fa32a2eeeb473422b129a16509e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-16 13:04:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"93ac617d40da51074d291f404702fe607db2d61d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-16 21:02:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e1056cd6ac54485403c93e87149768ff641bc35a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-04-16 21:46:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15cfc4fec8253f97b904c879f05e1f0f7d4eeffa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 21:47:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1796f3f1bd4b27414d0b092dccb5c7657c624eb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 22:58:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32e1cdee6dd1438980f9701d575e90a1\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e96c3a6e5c1342dfb1e7b64ecdf92d59 : SUCCESS in 2m 13s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2187bbcf455d495ca77651ccacd77f3a : SUCCESS in 3m 12s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/134e19d4ae9f4ddd92762ebee84f69d3 : SUCCESS in 2m 31s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/33990e7153224a4e949211036b5a1a4c : SUCCESS in 8m 52s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70735d887dc74925960d350b0a80137a : SUCCESS in 5m 43s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/685e2e7ed6be4304924c84ae710c140e : FAILURE in 1h 05m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/198acbfa459241c49f1fc5010b879e65 : SUCCESS in 40m 24s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/249bd367f1684c75ad24f1f17002870f : SUCCESS in 54m 58s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea2d0881ef214cd893fdfb4c2962b624 : SUCCESS in 55m 48s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8851b13787a41e188af680498d1db0a : SUCCESS in 29m 43s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/83fef12d3a1742d7b17e196e8f2cda17 : SUCCESS in 19m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbb5952f1e324897ad53bff09328fde0 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"6320e8ad5d6b8946ca4234875891709f2319cb43","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-17 00:54:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9df60f1f56e460ec43224eb4abe0e0fc85376262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 03:38:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16616770abfd4550bf30c0bbf2e4ac16\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d8f2c9dab9a41ec9dc26836098a973c : SUCCESS in 1m 23s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12144a12cf9749ba88f28bbacb2b177b : SUCCESS in 1m 34s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5de2fa82808f4f1fab5d5dddeeaf8b34 : SUCCESS in 1m 30s\n- puppet-openstack-unit-7.35-centos-9-stream https://zuul.opendev.org/t/openstack/build/3762618f6fa04a10b37c479e57e73251 : FAILURE in 3m 22s\n- puppet-openstack-unit-8.11-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e12676f3a904139b41a2da8571a49e2 : FAILURE in 2m 38s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/161e6a77cbe64c81b351180c8b10f669 : SUCCESS in 4m 35s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ca9d22f9eeb41a1b9fbb6c83392b69f : SUCCESS in 56m 35s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54c227da1fe44ec3b7d438fda15fc2f5 : SUCCESS in 50m 14s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6edacafe86543159d666771fa5dfdf6 : SUCCESS in 35m 24s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffcb1b0f2c5741d7b4836953a28dbcc1 : SUCCESS in 30m 14s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8308d04da2c4fa1bd645701bb322c83 : POST_FAILURE in 1h 01m 17s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8314e2e29cf942558f8f84bd6b264070 : SUCCESS in 1h 00m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ef1b5897c2e48e39e54c45e30d7debb : SUCCESS in 1h 23m 08s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed8e2986aaac45b4935ff07b357e505e : SUCCESS in 58m 34s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8791a9f08e540a28f381b84969bccd6 : SUCCESS in 50m 25s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d619d183fd064b9b9da2c195e47329b3 : SUCCESS in 53m 06s\n- puppet-openstack-litmus-7.35-centos-9-stream https://zuul.opendev.org/t/openstack/build/4417ee5222594db880cdb449339ed645 : FAILURE in 5m 57s\n- puppet-openstack-litmus-8.11-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c4108b6ca4c438cb6fa95a49fbd47e2 : FAILURE in 3m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce7266a8a93e4b59932f73ed9decef97 : SUCCESS in 18m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae8431ab3fd74e16817817f833b47400 : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":1},{"id":"792b8d117b9d9c8d98477ff582214faeb827934a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-17 06:38:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2bacc8586959949c407d6f2b0d64288b33fa7ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 07:56:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9601b6bc7afa4bec821798cbbaeda681\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bf472184bb543bdad4028b06e5e41f8 : SUCCESS in 2m 34s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b947901561243ba9a895bd2e86f3f05 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c41938678014c3090891aa7461889c8 : SUCCESS in 2m 38s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/824ece8fedd74838858c5e9b104c58c4 : SUCCESS in 5m 05s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/202ce69a9b204aacbfb30366f3d36f5d : SUCCESS in 7m 02s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/393c3efc1cae4da187ea97d1858dc725 : SUCCESS in 6m 00s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6b56aeffa3343019effccfc4cf54460 : SUCCESS in 41m 02s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35f89095461f4427b7cc79d05b6d9b9d : SUCCESS in 51m 01s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19aa0206f42b44c8b33599e0db3e1285 : SUCCESS in 40m 25s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3030c173fe04c2e97737c9e88ea3756 : SUCCESS in 39m 46s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/004b7eb4178b464a8a72cb7a9a13b26c : SUCCESS in 34m 00s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a667a5deffc640c4a18bd169005eea81 : SUCCESS in 1h 03m 58s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/da1e0794420e45709cac3d09e6642206 : SUCCESS in 1h 16m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8c133d0b614406b94595734c368ba0e : SUCCESS in 46m 24s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b3a9d1ba9bf44ba8eeefd946e5ece1a : SUCCESS in 51m 40s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae163818c42944e994e82ee4c210fd28 : SUCCESS in 55m 36s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/65f498d7cebc4e91b2cb9f825c0f3b14 : SUCCESS in 20m 48s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51de0f3dea4f4a76a0556dc90bbb6aec : SUCCESS in 15m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63d836a75fe548308bda72974c6c4ba8 : SUCCESS in 17m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/974461d76c97488aa124c016c1790a40 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":1},{"id":"ac34b2a62319331b1b8d5ba267c05ee031295443","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 07:57:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff5fab4633876c8265846e7544e54cc1b839780b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 09:15:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3729d99415d49fda1947b0e7d11b50b\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e2a291ba7824b4790e65ea1eca8a6fd : SUCCESS in 2m 53s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a723a3e9fa745069b18601878f76253 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b01b4fcbd22345a8a28ecf3dc7816ab9 : SUCCESS in 2m 30s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc78d07ae4d4413cbd72ac9ebddb9eb0 : SUCCESS in 11m 13s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eeb93867de0e41e6a5ae472aab3e4d89 : SUCCESS in 9m 06s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a64c308e5c41463fa741b072bb5765d8 : SUCCESS in 1h 13m 05s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d78e22f7be84d6b99adde345a9cb61a : SUCCESS in 1h 06m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd7d3ded28e24e2ca21dcf1fff37f4a8 : SUCCESS in 25m 08s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a62d24662a3462ab94c76463303c90c : SUCCESS in 39m 52s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d15fda72cd814c4e8bcf6c55e1c27750 : SUCCESS in 28m 07s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/fde58f06c6984439a791fd1531a245e0 : SUCCESS in 16m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a1de7ddf775413fb508dbba55c89832 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9d1c26c63b159d66293143ff11dbfeab42c8d6e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 09:15:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0a101b3cbd0362a482a540353569e9595a46667d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 09:15:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b57cfcf50a1b4c60acae58f94296b766\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d5b39f4904245e8831624c4f255813e : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12a3303211cf75c96e0e1c526b85634a14f687b5","revisions":{"12a3303211cf75c96e0e1c526b85634a14f687b5":{"kind":"REWORK","_number":1,"created":"2025-04-16 11:54:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/947454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/54/947454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/54/947454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/54/947454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/54/947454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/54/947454/1"}}},"commit":{"parents":[{"commit":"90d291e7f7cdcdc401eb5dcdeb84b84e23a3e8c8","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/90d291e7f7cdcdc401eb5dcdeb84b84e23a3e8c8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-16 11:54:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-16 11:54:18.000000000","tz":540},"subject":"Remove watch_log_file","message":"Remove watch_log_file\n\nThe option was deprecated during the 2024.2 cycle[1]. Processes may\nfail to start when it is set to True[2] so supporting it is no longer\nuseful.\n\n[1] 31603ac6dacd0b47ec9f02d99bfdd0a7b683119c\n[2] b977cb9becbeaaa67bb663bd8db5aa6846e495b9\n\nChange-Id: Iff16deb64b2b2776aa5deab4d27247b8d58022ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/12a3303211cf75c96e0e1c526b85634a14f687b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/12a3303211cf75c96e0e1c526b85634a14f687b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
