)]}'
{"id":"openstack%2Fpuppet-keystone~954238","triplet_id":"openstack%2Fpuppet-keystone~master~Ie7a889992974d3ed1053ad14623270c6fe182365","project":"openstack/puppet-keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-18 10:55:00.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie7a889992974d3ed1053ad14623270c6fe182365","subject":"DNM: testing","status":"ABANDONED","created":"2025-07-07 12:33:37.000000000","updated":"2025-07-18 10:55:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6ef56a8005a0e4eb59af65f2f2dee54a2c13038d","_number":954238,"virtual_id_number":954238,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-07-07 13:52:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-07 12:54:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb83434a20bc26a46b3704c9a5eee2a6f83ef4cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-07 12:33:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93c9781815071ee4778e1b62ce3faf25b540d059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 12:54:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a17d25e55e8a4ad2afd8162e007b0521\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca244018ae634644be7257d9b97aa15c : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70af79e5bf074137bef3e5f94b317618 : SUCCESS in 1m 27s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bec8947d4144756a3cd6a500f273705 : SUCCESS in 3m 19s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/148b046bb3d244d9b2af2fff7f4b6435 : FAILURE in 15m 24s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16d786ed15414367aec1842667a6aedd : FAILURE in 7m 22s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b16d0ac96a0d444e894f21e760c4d0ed : FAILURE in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1fff6b408765638435def0bedb741cec55f1750","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-07 13:31:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c3e15234f2f25ce1ea3ff21e213f0620771078c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 13:52:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06cd9822d35a4e3e811fbcfa41d264a6\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45f1f0c8792e4d4bb21bf2a9b8e584a7 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78ff11eed6e243f19f415834f37d939f : SUCCESS in 2m 22s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44c8c538c0f74f5abc4b870ee0f07859 : SUCCESS in 3m 15s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/927871ff6b19425883461c43a84a6ac4 : FAILURE in 15m 06s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/987b717c2a2d4258bee842d8d527b39e : FAILURE in 7m 23s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbfb35932b1d40239a419e25015ab1ba : FAILURE in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ef56a8005a0e4eb59af65f2f2dee54a2c13038d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-18 10:55:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"12a4f74f936181805156bb8e4f909ad355571909","revisions":{"d1f4d72acdf1343d8f554275c30cbf8a4fa30f17":{"kind":"REWORK","_number":1,"created":"2025-07-07 12:33:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/954238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/38/954238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/38/954238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/38/954238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/38/954238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/38/954238/1"}}},"commit":{"parents":[{"commit":"9ee0886a2ca8b269e428df4655e217bb5a592c15","subject":"Drop unnecessary \"expect\" layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/9ee0886a2ca8b269e428df4655e217bb5a592c15"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-07 12:33:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-07 12:33:28.000000000","tz":540},"subject":"DNM: testing","message":"DNM: testing\n\nChange-Id: Ie7a889992974d3ed1053ad14623270c6fe182365\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d1f4d72acdf1343d8f554275c30cbf8a4fa30f17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d1f4d72acdf1343d8f554275c30cbf8a4fa30f17"}]},"branch":"refs/heads/master"},"12a4f74f936181805156bb8e4f909ad355571909":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-07 13:31:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/954238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/38/954238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/38/954238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/38/954238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/38/954238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/38/954238/2"}}},"commit":{"parents":[{"commit":"57ce8e678194499877fbc95f09079aef82c7101f","subject":"Drop unnecessary \"expect\" layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/57ce8e678194499877fbc95f09079aef82c7101f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-07 12:33:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-07 13:31:36.000000000","tz":0},"subject":"DNM: testing","message":"DNM: testing\n\nChange-Id: Ie7a889992974d3ed1053ad14623270c6fe182365\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/12a4f74f936181805156bb8e4f909ad355571909"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/12a4f74f936181805156bb8e4f909ad355571909"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
