)]}'
{"id":"openstack%2Fpuppet-keystone~958977","triplet_id":"openstack%2Fpuppet-keystone~master~I40795145be181c433763e24f849cfc243c41c31f","project":"openstack/puppet-keystone","branch":"master","hashtags":[],"change_id":"I40795145be181c433763e24f849cfc243c41c31f","subject":"Fix lint error","status":"MERGED","created":"2025-08-31 06:07:53.000000000","updated":"2025-08-31 09:09:38.000000000","submitted":"2025-08-31 09:09:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958977","meta_rev_id":"e1129fb0074b1090e0c49ba9a1c3d7c9ffaabb17","_number":958977,"virtual_id_number":958977,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-31 09:09:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-08-31 07:41:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-08-31 07:41:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-31 07:37:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f90b264ab15c13e9ee28be76459c0df5eff1e311","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-31 06:07:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f387a56d54b3e9efaf372a8adb0a5d74d9f6429a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-31 07:37:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6c0dc0cba6e4b4680e01266641be9e3\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a033ccf70a094d498c4cceaf12833232 : SUCCESS in 5m 14s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41c48a3b2bca4d889c225c953cfcc86a : SUCCESS in 2m 40s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/58f69d90f1da4bc4a16382d62f2b647a : SUCCESS in 5m 50s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f42bbb64f186420c89dfc6f6cd03019c : SUCCESS in 9m 47s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0312970e21d8403ab1cfdce721e94480 : SUCCESS in 4m 42s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e65d3d80634493da710aa46b4a9e6d5 : SUCCESS in 26m 39s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66cb36d286e04656a54b6487450443d9 : SUCCESS in 1h 02m 15s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/439cbb08e1d84dbbb6b173d1653b574e : SUCCESS in 23m 45s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c64121760e61408881da29198eb7e4ec : SUCCESS in 41m 56s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cfa508f8bf640228d3e7d5a498a2f19 : FAILURE in 25m 36s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/43f17cd7ae03420a8c0437be40b83f0d : SUCCESS in 1h 09m 43s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d137a9d7275a4210bd8004882a2d0c7f : SUCCESS in 57m 53s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fdb7e1cd040340b280aa58d758e1b020 : SUCCESS in 59m 33s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aad14c4d4fcb4d68953651853f79e561 : SUCCESS in 41m 45s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4dfba771fb174c3ab2e23f7210a6ffcf : SUCCESS in 40m 46s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2fcb06bee9c4de3990f67e28dc82481 : SUCCESS in 1h 07m 14s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb7b65e80bd740fa95d2f62d3436170b : SUCCESS in 1h 16m 25s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5932097bbb243a5988f6e12ad3099d4 : SUCCESS in 56m 51s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc06a15f85e74822ac7ef2814442081a : SUCCESS in 56m 10s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c15834f51d84209a203fb32c90d087c : SUCCESS in 44m 20s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/aecddaf30e794709b68f385ae9e3eae7 : SUCCESS in 23m 58s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7ba84e0f4a04e16a8528b0772ebfa81 : SUCCESS in 22m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c35ea569c9f248398513edfe473ad8e8 : SUCCESS in 18m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"580e0c52b93b47021e308042faceba07b59667d7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-31 07:41:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22fe636e68170e9c33ad7a0246a41510cf6ca02e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-31 07:42:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"159562f5d4c700c1e3a9cfa44b72c55ebf70dc5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-31 09:09:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/378e8530d80445288d44c4261081da8f\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f194891c3d9a4ba082de163f8df872fa : SUCCESS in 2m 09s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b02169b500224002b5d3045a29a16728 : SUCCESS in 2m 51s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/7d5fa536650f447bab1c2266d6de09c9 : SUCCESS in 8m 50s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5177db68acd14e2f9464e176c70a425b : SUCCESS in 8m 45s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6b126e8e97043bc88dae750f764a6a6 : SUCCESS in 1h 07m 04s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a094196e6b149bfa9fc1095d13a18e4 : SUCCESS in 1h 23m 41s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba7a76b3bcf94d3e9b9406c5bc1406d6 : SUCCESS in 1h 05m 13s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/40b099a0728a46c08e610f524a9c721f : SUCCESS in 52m 00s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e72f58a06554433fb26306362bd4b7c5 : SUCCESS in 57m 20s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/caa25267f3c24ad3bf8ac86c33c5fa51 : SUCCESS in 25m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e1129fb0074b1090e0c49ba9a1c3d7c9ffaabb17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-31 09:09:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24ac6766b8b31715a7f7b0a1d11ac4623ee8d76e","revisions":{"24ac6766b8b31715a7f7b0a1d11ac4623ee8d76e":{"kind":"REWORK","_number":1,"created":"2025-08-31 06:07:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/958977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/77/958977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/77/958977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/77/958977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/77/958977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/77/958977/1"}}},"commit":{"parents":[{"commit":"394b1ccbf1d976f46fad2b391d4f8b7881f2e478","subject":"Merge \"Replace wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/394b1ccbf1d976f46fad2b391d4f8b7881f2e478"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-31 06:07:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-31 06:07:45.000000000","tz":540},"subject":"Fix lint error","message":"Fix lint error\n\nFix the following error detected in CI.\n\nmanifests/wsgi/apache.pp - ERROR: there should be a single space before\nan opening brace on line 182\n(check: manifest_whitespace_opening_brace_before)\n\nChange-Id: I40795145be181c433763e24f849cfc243c41c31f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/24ac6766b8b31715a7f7b0a1d11ac4623ee8d76e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/24ac6766b8b31715a7f7b0a1d11ac4623ee8d76e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
