)]}'
{"id":"openstack%2Fpuppet-keystone~967726","triplet_id":"openstack%2Fpuppet-keystone~master~I81b06e1ac8539cefadcb7566c8ad9a7b2e7413b4","project":"openstack/puppet-keystone","branch":"master","topic":"hashclient-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-28 13:44:54.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I81b06e1ac8539cefadcb7566c8ad9a7b2e7413b4","subject":"cache: Follow renaming of hashclient parameters","status":"MERGED","created":"2025-11-19 18:00:38.000000000","updated":"2025-11-28 16:37:21.000000000","submitted":"2025-11-28 16:36:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967726-hashclient-opts","meta_rev_id":"935885c2969a72f5452eb499de88c952d4113283","_number":967726,"virtual_id_number":967726,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-28 16:36:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-11-26 15:03:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-11-28 08:39:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-11-28 08:39:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-19 19:46:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-26 15:03:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-11-28 08:39:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"1d8a069bef7ffb658fa8810da93589cf3f6ce7ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-19 18:00:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f39850604546604ef881868f52c4645e3da29180","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-19 18:08:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e900f4f554d3589d5b843b0fc60d957c4f784874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 19:46:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5e27de516414e1cb9aba17e88a184e3\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fc76d82fe2a404e90a0b21e8c1d2987 : FAILURE in 2m 51s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3823fd5ee1d2425b9f24e3c7851044a6 : SUCCESS in 2m 54s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/4808c8a87a434d018253d4e7672f9626 : FAILURE in 6m 26s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcfd2f9b1c8b432987fd744c20e37049 : FAILURE in 12m 06s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/145272472503441ebd32ba47f20c98dc : FAILURE in 14m 47s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40e384abdff644819b7c7b0174bef659 : SUCCESS in 1h 01m 29s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d55a9eb908149bb9a14a598151d1dac : SUCCESS in 1h 12m 32s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7071bcab12984d538eb0450b4ab3e448 : FAILURE in 30m 22s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26881f8f4b3c417c985139e377d03476 : SUCCESS in 43m 10s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03e7c7eded2c4c8ca89f9c7c52105041 : FAILURE in 48m 20s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e47316ecabc949feb7915c8b5af163ba : SUCCESS in 1h 07m 10s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b4790123ee446309239cc1e7c58cb42 : SUCCESS in 41m 01s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a7f1760948f49cdacf35d40fe14170d : SUCCESS in 1h 00m 39s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/399915548453485f9e370f39ed68fb4c : SUCCESS in 51m 50s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/00b7a0b5940b454da86897f473150e3e : SUCCESS in 47m 38s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ffd363cb9024bcfb5ed1be4e8f41934 : SUCCESS in 33m 25s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/80c667c28a8d4c75b89c716487871e27 : SUCCESS in 38m 33s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e33d94273ed471d84271a234090f1c5 : SUCCESS in 58m 38s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5f9812423864b919ad06a95183860f0 : SUCCESS in 55m 17s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ea6d384ded443a586f397ac32066c7a : SUCCESS in 43m 35s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/54f3d662cc97426b82e7d18495ea3cac : SUCCESS in 24m 31s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bffb326499d43eb83a3ddc60b11d972 : TIMED_OUT in 1h 32m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6578f3be33ce4668ba067ab9036b7e9b : TIMED_OUT in 1h 32m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dc72a5b270f40b485288f320be758a5 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"c431f275d05531f5fd40958d286c8a0d4f31c43e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-20 11:17:33.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e4aa6dfdd62231b3e0288d502cf6b8437a8f3cbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-20 11:18:49.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"307846588a237f1a103c437b474cc51247e58cfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 12:55:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84a580b1661d4f079aee1c9ad83e1c24\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09750f47fd724eb699c65af88b02da62 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3510ab9e6211469cba86b60865e0b7b6 : SUCCESS in 2m 42s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/4d45d6c3789d48b78cac2a938544bda2 : SUCCESS in 11m 47s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/780b6e87b1ac4d57a7538f4ad49bb692 : SUCCESS in 10m 25s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c73559bcce93423886ba382917105aa3 : SUCCESS in 5m 54s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e3d017619304deab4b6efa02b4f4625 : SUCCESS in 58m 51s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8faf77bb9fd24bada1ada7bc6f1903d4 : SUCCESS in 34m 20s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcd0c5de8d78456f984f6f181b80abe3 : FAILURE in 31m 51s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/426e4302e6be4ff7a705c03273ad8440 : SUCCESS in 45m 57s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d67ef29551b143c69a3d2eda9a994a81 : FAILURE in 50m 30s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d15031018dd48769f33da0c1c42c978 : FAILURE in 1h 28m 00s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/105560dbaf3a49ce8c2c7a3a08b9f8b6 : SUCCESS in 1h 21m 54s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3a0ce470a114f47984fa3ff28f3d5fb : SUCCESS in 1h 01m 31s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4dea4b82f6084c42b2180d98ad24b336 : SUCCESS in 53m 37s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/699bdb762e46426eabc0e4fbfaa09b81 : SUCCESS in 55m 01s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc13c2b499d74a8b958598814f65980a : FAILURE in 1h 25m 26s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4f95913eacd446b883819290bc4017f : SUCCESS in 1h 16m 34s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7ef4d691df74a5c8b34ee93ee91332f : SUCCESS in 59m 23s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3bdeb11a6b05418db32d90890135c3ac : SUCCESS in 56m 02s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c75e3fcf4b3147f895fc0b7d7c6f43f8 : SUCCESS in 53m 39s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f848d032cc7461ea72508261cf1af95 : SUCCESS in 24m 50s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d74f83290cf44694bf9d565e73096de0 : TIMED_OUT in 1h 31m 56s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56326b023a69414f98cbe0b94ef5157c : TIMED_OUT in 1h 32m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a3750e9bd074b81b38404131595a00c : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":4},{"id":"3234663bc5a843bef48e59ae89d8ee66da309cd9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-22 15:41:09.000000000","message":"Topic set to hashclient-opts","accounts_in_message":[],"_revision_number":4},{"id":"93bf88958a6a16e4955869d6dadabc7f88dc396c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-25 04:08:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3fcc95b60741cfab2e7b4597a022d20ffbfd346d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 05:42:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/734402a499f54584b31b198934f7ca50\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/591f4358d6fd43239534eeae197b03c9 : SUCCESS in 2m 08s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44340036b0c443489b1154097a412b01 : SUCCESS in 2m 58s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/1bb194c8faea4a0081b178da67e854cf : SUCCESS in 7m 42s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70326a96987544818b810d0814f589d7 : SUCCESS in 14m 34s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c6f2ee1de1f4fbea13ad7461ed0934f : SUCCESS in 13m 13s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47ee554fbecb4f56b4c814b43924d716 : SUCCESS in 58m 25s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42b73f8c1cb3454db18c5f95528d40f2 : SUCCESS in 1h 11m 30s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/489f2f3e1caa41c9aedae882747de260 : SUCCESS in 51m 19s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b34123a143c4437a95afb9cb1206a690 : SUCCESS in 23m 34s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/575904a4c3334220a45c656d602f8c43 : FAILURE in 44m 42s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/601fef2c491f4ee5865e3d3f619318cc : SUCCESS in 1h 06m 13s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7667e0f020e4e6ab2b91934dd3364ac : SUCCESS in 1h 23m 34s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c41baaaff83c47b58c2d4238578e6ffb : SUCCESS in 49m 08s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a3aabcdb7044747a0a2a0ccd03067f8 : SUCCESS in 58m 06s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7615f26bcfab4dfd96c25d1893e6f7c6 : SUCCESS in 1h 05m 39s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6afe5e43ff74c69a1c8b7d72d68b023 : SUCCESS in 1h 08m 50s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b1df7356e474b65ba92a175c506e8d6 : SUCCESS in 1h 24m 33s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1aa680a39493446ba82a913cdb8c27e1 : SUCCESS in 1h 04m 17s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae7cea96da3f4cb5ac3c0a1f7280f2c8 : SUCCESS in 30m 21s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed123fa353ee49f69bd558dd2e3480fd : SUCCESS in 1h 03m 41s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/a65ce5c8317c4900af5e57baaea98c38 : SUCCESS in 27m 25s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/826449e4781c448788c2d76f329794bd : TIMED_OUT in 1h 32m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca27281007974a1f9cc131046d03c908 : TIMED_OUT in 1h 31m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4c1a59f18484f12a4645164b27da9c8 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":4},{"id":"686a7d01fda5997c66ddaa3eca78f529e71b00d6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-11-26 15:03:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5e37c44cc3c84928921d312db6c0bf22656434a1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-11-28 08:39:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"195a5578f331f382413115e410942f99cae00b9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 08:39:11.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75b9bd7489134fa8a018539e67704eb2\n\n\nWarning:\n  Change 967710,5 in project openstack/puppet-oslo does not share a change queue with 967726 in project openstack/puppet-keystone","accounts_in_message":[],"_revision_number":4},{"id":"b54b4d9a1dcafd8c5959ea5152aeff73283b684c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-28 13:44:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7a0b5ace1f6cce08bc0d1c6aeaf4873bbf347027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 15:19:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/461037309a124aaf9ca9c5967aab411d\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2685d730d7034ae2969da15376a2c1e1 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5672da109c8e485eabb88e5e95a79409 : SUCCESS in 3m 04s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/2b530e9c411b47efaa68a461e3254a4a : SUCCESS in 11m 08s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81f77b3c257d4655bc41c7345e344f2a : SUCCESS in 13m 14s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab34d4544047410f98ccf4634edb64cb : SUCCESS in 9m 36s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdd1e808a30e4ec5a2f26817a9f9ba7b : SUCCESS in 53m 58s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c6412da14864742b9eedabc37d9f81b : SUCCESS in 1h 26m 03s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/757c3c3da9b2462e9edad4498cb472bc : SUCCESS in 50m 48s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/287a19afc46f49398424b2e2cf594003 : SUCCESS in 44m 35s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da1a3a6e8aea4652bd01fb196dc46b6c : FAILURE in 52m 03s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/73337efa7708411089910c20a06be336 : SUCCESS in 1h 12m 17s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ac5c6cc3cab44eaa61d680cf8c47339 : SUCCESS in 1h 22m 31s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d366fc585654a429887e01e55f65837 : SUCCESS in 1h 02m 21s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8747c2e490e741dbb4edac7aa0645487 : SUCCESS in 54m 06s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9532f62e48a404eba42f336e8825f94 : SUCCESS in 59m 26s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9bc5d8adb414ba0bec5ef93a22e1969 : SUCCESS in 1h 17m 12s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb74f62db9944bd2ac5490c7cdf76903 : SUCCESS in 1h 12m 35s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/105a402442974b208eb5da425f3af6b7 : SUCCESS in 57m 58s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/acac28d456ba4903bb65413f01da18bc : SUCCESS in 54m 23s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f42ad9f62c147ad96b9c916f1c6233f : SUCCESS in 57m 01s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/98a76cc63fe04cf0aae4789d9e7ea82c : SUCCESS in 24m 18s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b593e960db64472985554c226696ba22 : SUCCESS in 9m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/965b63dff43d456890b44a16a0cb5ac3 : SUCCESS in 22m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8982eedb3b6041fa8b57dd9ec828568a : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":4},{"id":"d548d012a3d0719aa82e4836106eba5630503fb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 15:19:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f1515ef1ba0e5b746cfecb2e96f475322abb4e87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 16:36:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5c0cd8a830d4d9d8baff3d99d0f4c41\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/132c16bfaf48487c810ed4424d15e101 : SUCCESS in 2m 56s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c74d78ff2d14d41a6c5e3dbd25a0439 : SUCCESS in 2m 59s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/1a8c9c2b086b4b808b0a37f10dcf40e4 : SUCCESS in 6m 32s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cecdd97169d4588a97fd9b7163b0516 : SUCCESS in 6m 07s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c192c760c1046868250a34f04ad31bf : SUCCESS in 1h 04m 34s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/17a1656443d24a6d833da7100759dccd : SUCCESS in 1h 15m 21s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0de721539504e50a2f5664cf46639ef : SUCCESS in 43m 31s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdf53fedc85c4cf0bd8625dd94d6fbd5 : SUCCESS in 54m 00s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8df5f87bc796460b8ab2c94e5cbb3770 : SUCCESS in 44m 38s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a3d5468ceaa4b94a8e79cfea4adbadf : SUCCESS in 18m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66dcc615503a4eec8c32b05ae1e2a1a1 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":4},{"id":"4b8b33c29db62dbd9e6033c1d5eb2e46e3091753","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 16:36:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"935885c2969a72f5452eb499de88c952d4113283","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 16:37:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18346f7613af4f22bafe7a73a856f201\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5137bfd151bd489e888f305dec215b26 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"39febbf9fe4649c3226b5360f9106fcd33f6beab","revisions":{"35ed872091174832260080b9c51135ae68ce5161":{"kind":"REWORK","_number":1,"created":"2025-11-19 18:00:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/967726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/26/967726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/1"}}},"commit":{"parents":[{"commit":"d62d032175dfd5ae3bc4c663554086f2514c54c9","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d62d032175dfd5ae3bc4c663554086f2514c54c9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-19 17:59:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-19 18:00:30.000000000","tz":540},"subject":"cache: Follow renaming of hashclient parameters","message":"cache: Follow renaming of hashclient parameters\n\nDepends-on: https://review.opendev.org/967580\nDepends-on: https://review.opendev.org/967710\nChange-Id: I81b06e1ac8539cefadcb7566c8ad9a7b2e7413b4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/35ed872091174832260080b9c51135ae68ce5161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/35ed872091174832260080b9c51135ae68ce5161"}]},"branch":"refs/heads/master"},"eb490eb35577c5fdcb997d10d1750deb0f691808":{"kind":"REWORK","_number":2,"created":"2025-11-19 18:08:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/967726/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/26/967726/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/2"}}},"commit":{"parents":[{"commit":"d62d032175dfd5ae3bc4c663554086f2514c54c9","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d62d032175dfd5ae3bc4c663554086f2514c54c9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-19 17:59:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-19 18:07:58.000000000","tz":540},"subject":"cache: Follow renaming of hashclient parameters","message":"cache: Follow renaming of hashclient parameters\n\nDepends-on: https://review.opendev.org/967580\nDepends-on: https://review.opendev.org/967710\nChange-Id: I81b06e1ac8539cefadcb7566c8ad9a7b2e7413b4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/eb490eb35577c5fdcb997d10d1750deb0f691808"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/eb490eb35577c5fdcb997d10d1750deb0f691808"}]},"branch":"refs/heads/master"},"b86edd47b10e0fc76ea9bd5cb1807bcc9cc940e4":{"kind":"REWORK","_number":3,"created":"2025-11-20 11:17:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/967726/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/26/967726/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/3"}}},"commit":{"parents":[{"commit":"d62d032175dfd5ae3bc4c663554086f2514c54c9","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d62d032175dfd5ae3bc4c663554086f2514c54c9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-19 17:59:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-20 11:17:33.000000000","tz":0},"subject":"cache: Follow renaming of hashclient parameters","message":"cache: Follow renaming of hashclient parameters\n\nDepends-on: https://review.opendev.org/967580\nDepends-on: https://review.opendev.org/967710\nChange-Id: I81b06e1ac8539cefadcb7566c8ad9a7b2e7413b4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/b86edd47b10e0fc76ea9bd5cb1807bcc9cc940e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/b86edd47b10e0fc76ea9bd5cb1807bcc9cc940e4"}]},"branch":"refs/heads/master"},"39febbf9fe4649c3226b5360f9106fcd33f6beab":{"kind":"REWORK","_number":4,"created":"2025-11-20 11:18:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/967726/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-keystone","ref":"refs/changes/26/967726/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-keystone refs/changes/26/967726/4"}}},"commit":{"parents":[{"commit":"d62d032175dfd5ae3bc4c663554086f2514c54c9","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/d62d032175dfd5ae3bc4c663554086f2514c54c9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-19 17:59:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-20 11:18:49.000000000","tz":0},"subject":"cache: Follow renaming of hashclient parameters","message":"cache: Follow renaming of hashclient parameters\n\nDepends-on: https://review.opendev.org/967580\nDepends-on: https://review.opendev.org/967710\nChange-Id: I81b06e1ac8539cefadcb7566c8ad9a7b2e7413b4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/39febbf9fe4649c3226b5360f9106fcd33f6beab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-keystone/commit/39febbf9fe4649c3226b5360f9106fcd33f6beab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
