)]}'
{"id":"openstack%2Fpuppet-magnum~749621","triplet_id":"openstack%2Fpuppet-magnum~master~I59ce740ad85967e181da328e11d394cc1c029ab6","project":"openstack/puppet-magnum","branch":"master","topic":"749621","hashtags":[],"change_id":"I59ce740ad85967e181da328e11d394cc1c029ab6","subject":"Add oslo_middleware/enable_proxy_headers_parsing option","status":"MERGED","created":"2020-09-03 02:23:27.000000000","updated":"2020-09-03 18:03:58.000000000","submitted":"2020-09-03 17:59:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749621-1599155977435-b29784c5","meta_rev_id":"ccb56c9c1b3af3128457a629a919b9ba4a4183af","_number":749621,"virtual_id_number":749621,"owner":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-03 17:59:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-03 16:13:12.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-03 12:53:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-03 16:13:12.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-03 12:53:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-09-03 16:13:12.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-03 17:59:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"03974a6068d56c61d9456c9f46184ab9b32dc5b9","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2020-09-03 02:23:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0deca11563f6d960826d1df6980deb5616cdf70b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 03:43:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4723f243302a410ca41b3ce159bb6f00 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85a75e1b6fb24e63a241613b055d2cc2 : SUCCESS in 2m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01ff518f922a4ad393343f70006494fa : SUCCESS in 2m 35s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f284f73cb3de40119bd40eb14db9a553 : SUCCESS in 11m 44s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5cf98632493043c7af297f08d3df49bd : SUCCESS in 9m 03s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8e623bc2f1241f58dec3e4f7fbeeec4 : SUCCESS in 7m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec57780c082a4c0582878baec48b1059 : SUCCESS in 7m 35s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c05faddb6c34c4d8aa6211ad6008b0d : SUCCESS in 54m 02s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7443f8763b234805baacbd2bb5f24678 : SUCCESS in 1h 15m 18s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/a481280d4aaf4ab0af779fbf346dd630 : SUCCESS in 17m 04s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1082d1a706048329600acd17688b306 : SUCCESS in 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"656b1822894bda47f2d21ef0cf316cb5efc59ca5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-03 03:52:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nNeeds some improvements in test codes.\n\nAlso, could you add a release note about the new parameter ?","accounts_in_message":[],"_revision_number":1},{"id":"2f61faf55e36f5cf81b3659848c708e0f4d00256","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2020-09-03 05:26:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"323d4cdb2ff157c7c329894fd1f6b0bfe35499ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 07:11:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97b02a28a6c6409fb529f6d891e4eec2 : SUCCESS in 2m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d920754191fd42388b8aee17ce384c01 : SUCCESS in 2m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/513d7f597b67405f912d87becbb6261c : SUCCESS in 2m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/eabc0ee8dc5940c3931eb4a5a63acda1 : SUCCESS in 11m 53s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/378382779a0e4462a4782a859d258c3b : SUCCESS in 8m 51s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53c980cc2c824bb4b85df8925b374007 : SUCCESS in 6m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16370fddc7564226b420715b735c322a : SUCCESS in 8m 41s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efd9e8d1746b49918f4fd37d55d2d540 : SUCCESS in 48m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f64731205a914b8083ecd85e1db18b24 : SUCCESS in 1h 37m 21s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/65d08a75202c4850946f5cb2decbc6bf : SUCCESS in 17m 24s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56bf8b9dd3064f289a21c47759841888 : SUCCESS in 15m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"94ddc0d4fe8b94bcaa9bfb6086f03ce008424532","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2020-09-03 08:10:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"130883ad9850054a2af8a6b3d12a3b4cdb83d95a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 09:42:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23eee38a11224989a8a02117effe6e45 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5577e33b11dd47b490dfb434b065c8c7 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e67d31ad1d404dbd900a14aa71260db7 : SUCCESS in 3m 01s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/180af0c49fad47fb98467a9a43cb9cc6 : SUCCESS in 11m 09s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/04085e804452441ea6648dafd9ad983f : SUCCESS in 8m 45s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a121f510cd7411fab08a7a50d9802d9 : SUCCESS in 7m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0689b30082bc4e8d9eb5f9b02430c070 : SUCCESS in 8m 42s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bcdb1c2456f64b0fb77381397c7a6135 : SUCCESS in 44m 52s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a6fd00ebfcc24cb8a49a36ea26f13b10 : SUCCESS in 1h 24m 05s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/7831f21590914ad0a8fdbda1c599e441 : SUCCESS in 20m 18s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34a5f422668d4f58bfb01840e3ca8acf : SUCCESS in 14m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffe51ca8e03a47e8beb275035c5eeeb3 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":3},{"id":"6ca22d97355e66411d19a478f4373d5eb525ff82","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-03 10:47:15.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nThanks for addressing my comments.\nI think the current release note can be improved a bit, but the other things look good to me.","accounts_in_message":[],"_revision_number":3},{"id":"7e83cad02071a84691672be6fc56ac5368abd799","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2020-09-03 12:37:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a2c214dbd633343c503fde121e20ca50d00626a3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-03 12:53:57.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4e3829d5f156279824e11e4a7b1801d587aaa2e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 14:12:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e82ea36357814de5b6b4de28f8224730 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25622c9c14b4478780ba1f27b117d862 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca17f7eca30542d18cbb1b1e996481d1 : SUCCESS in 3m 11s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/bbeed07ae93e4fff8c1b6573c06c6fc6 : SUCCESS in 12m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b8a35c232bff438bbc292cc81db26448 : SUCCESS in 12m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8993db2efbd4ce79ae28e9127e89a17 : SUCCESS in 11m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55a18a8f0cfb4dc09f29613d1b9560ac : SUCCESS in 13m 21s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93b45db206c94afead8ead8709835a08 : SUCCESS in 47m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0e52ff6fa65f4e10ae80b231a425c19c : SUCCESS in 1h 20m 12s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/40db5c4a174d415099a7f9e966709ef4 : SUCCESS in 24m 24s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72bc0e534f7e47239d760dd855dfd354 : SUCCESS in 18m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37c72c456e0c4fc58db17070c7f34fdb : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":4},{"id":"71ce27d7fb7e50fb2adbadff0cecdf20902d6cbc","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-03 16:13:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f6d0b04213841404a17dc4504f13597b6583156d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 16:13:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"916f3d63fbc103e08e815d9e2d6562e217325440","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 17:59:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"1b1459f9454e3001eb25a908219b04fc092be261","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 17:59:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f15ad7997ea4d48bf5c1a7f6ed79351 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66944a0345a24e9eb4986a13bc2f2a02 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2769462b9144d25a8a3ef4de692f04e : SUCCESS in 3m 36s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/32b711ade01946cd952f77ed3135376d : SUCCESS in 13m 24s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9a0986ccb5334e1abaf604b7c57eef59 : SUCCESS in 10m 11s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f71989af0e74c9e8b8618f0ecc1b15f : SUCCESS in 8m 24s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5e38cd88433c46cb8346f6b3906eaaae : SUCCESS in 1h 29m 00s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/697c60da36a54c46b84c0faafe6cc537 : SUCCESS in 23m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b12c1b9e52f472d871551d0d9be4958 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":4},{"id":"4647526c7d23146f3d2e323abd0c6ad616884e97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 18:03:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40fd46423ee74191815c3f800d935cb8 : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e2b05c8956cda43ce73559750eaabaf640b7a8d1","revisions":{"6a867571fabc54c8bf03a5569dc80945e6a17541":{"kind":"REWORK","_number":1,"created":"2020-09-03 02:23:27.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/21/749621/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-magnum","ref":"refs/changes/21/749621/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/1"}}},"commit":{"parents":[{"commit":"f66eea46e29edca1cb304bbb43bd6e589ac1ace6","subject":"Fix testing for magnum::keystone::authtoken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-magnum/commit/f66eea46e29edca1cb304bbb43bd6e589ac1ace6"}]}],"author":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2020-09-02 08:07:35.000000000","tz":600},"committer":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2020-09-02 23:58:27.000000000","tz":600},"subject":"Add oslo_middleware/enable_proxy_headers_parsing option","message":"Add oslo_middleware/enable_proxy_headers_parsing option\n\nChange-Id: I59ce740ad85967e181da328e11d394cc1c029ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-magnum/commit/6a867571fabc54c8bf03a5569dc80945e6a17541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-magnum/commit/6a867571fabc54c8bf03a5569dc80945e6a17541"}]},"branch":"refs/heads/master"},"f185058fdb3a4c712161b084989578c5fa083017":{"kind":"REWORK","_number":2,"created":"2020-09-03 05:26:57.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/21/749621/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-magnum","ref":"refs/changes/21/749621/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/2"}}},"commit":{"parents":[{"commit":"f66eea46e29edca1cb304bbb43bd6e589ac1ace6","subject":"Fix testing for magnum::keystone::authtoken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-magnum/commit/f66eea46e29edca1cb304bbb43bd6e589ac1ace6"}]}],"author":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2020-09-02 08:07:35.000000000","tz":600},"committer":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2020-09-03 05:23:25.000000000","tz":600},"subject":"Add oslo_middleware/enable_proxy_headers_parsing option","message":"Add oslo_middleware/enable_proxy_headers_parsing option\n\nChange-Id: I59ce740ad85967e181da328e11d394cc1c029ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-magnum/commit/f185058fdb3a4c712161b084989578c5fa083017"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-magnum/commit/f185058fdb3a4c712161b084989578c5fa083017"}]},"branch":"refs/heads/master"},"f4348cb05040f33e224eb736cedf9d9f36d5adc5":{"kind":"REWORK","_number":3,"created":"2020-09-03 08:10:57.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/21/749621/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-magnum","ref":"refs/changes/21/749621/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/3"}}},"commit":{"parents":[{"commit":"f66eea46e29edca1cb304bbb43bd6e589ac1ace6","subject":"Fix testing for magnum::keystone::authtoken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-magnum/commit/f66eea46e29edca1cb304bbb43bd6e589ac1ace6"}]}],"author":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2020-09-02 08:07:35.000000000","tz":600},"committer":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2020-09-03 08:09:59.000000000","tz":600},"subject":"Add oslo_middleware/enable_proxy_headers_parsing option","message":"Add oslo_middleware/enable_proxy_headers_parsing option\n\nChange-Id: I59ce740ad85967e181da328e11d394cc1c029ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-magnum/commit/f4348cb05040f33e224eb736cedf9d9f36d5adc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-magnum/commit/f4348cb05040f33e224eb736cedf9d9f36d5adc5"}]},"branch":"refs/heads/master"},"e2b05c8956cda43ce73559750eaabaf640b7a8d1":{"kind":"REWORK","_number":4,"created":"2020-09-03 12:37:37.000000000","uploader":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"ref":"refs/changes/21/749621/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-magnum","ref":"refs/changes/21/749621/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-magnum refs/changes/21/749621/4"}}},"commit":{"parents":[{"commit":"f66eea46e29edca1cb304bbb43bd6e589ac1ace6","subject":"Fix testing for magnum::keystone::authtoken","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-magnum/commit/f66eea46e29edca1cb304bbb43bd6e589ac1ace6"}]}],"author":{"name":"Jake Yip","email":"jake.yip@ardc.edu.au","date":"2020-09-02 08:07:35.000000000","tz":600},"committer":{"name":"Jake Yip","email":"jake.yip@unimelb.edu.au","date":"2020-09-03 12:37:26.000000000","tz":600},"subject":"Add oslo_middleware/enable_proxy_headers_parsing option","message":"Add oslo_middleware/enable_proxy_headers_parsing option\n\nChange-Id: I59ce740ad85967e181da328e11d394cc1c029ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-magnum/commit/e2b05c8956cda43ce73559750eaabaf640b7a8d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-magnum/commit/e2b05c8956cda43ce73559750eaabaf640b7a8d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
