)]}'
{"id":"openstack%2Fpuppet-manila~662974","triplet_id":"openstack%2Fpuppet-manila~stable%2Frocky~I31fff6a1b6fede860b8aaee0079183924e8b31d6","project":"openstack/puppet-manila","branch":"stable/rocky","topic":"bp/manila-external-ganesha","hashtags":[],"change_id":"I31fff6a1b6fede860b8aaee0079183924e8b31d6","subject":"Add parameters necessary to use a remote Ganesha service","status":"MERGED","created":"2019-06-04 08:45:23.000000000","updated":"2019-06-05 02:05:45.000000000","submitted":"2019-06-05 02:05:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662974-1559700345009-4bf1be21","meta_rev_id":"694c5e5f8a2b69545414e1170369b2fcfd68fa6e","_number":662974,"virtual_id_number":662974,"owner":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-06-04 14:40:47.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":2,"date":"2019-06-05 02:05:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-06-05 01:09:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-06-04 17:16:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-06-05 01:09:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 17:16:05.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-05 01:09:16.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-06-05 02:05:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e088334f09400f451163879871c54cef48e4f0d","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-06-04 08:45:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7c98ee532f45facec5b56062d66b92a8b2cbe740","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-06-04 08:46:04.000000000","message":"Topic changed from bp/manila-external-ganesha-stable/rocky to bp/manila-external-ganesha","accounts_in_message":[],"_revision_number":1},{"id":"761d74644eab6739a8e3df87d317941c6d7abb2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 10:22:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/74/662974/1/check/puppet-openstack-lint/52a3dd8/ : RETRY_LIMIT in 1m 21s\n- puppet-openstack-syntax-4 http://logs.openstack.org/74/662974/1/check/puppet-openstack-syntax-4/cbb3a3e/ : SUCCESS in 5m 36s\n- puppet-openstack-syntax-5 http://logs.openstack.org/74/662974/1/check/puppet-openstack-syntax-5/490d9e6/ : SUCCESS in 6m 27s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/74/662974/1/check/puppet-openstack-unit-4.8-centos-7/d6aca77/ : SUCCESS in 16m 22s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/74/662974/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/9193ce2/ : SUCCESS in 46m 11s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/74/662974/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/02cc1e9/ : SUCCESS in 49m 07s\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/74/662974/1/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/32d74ec/ : TIMED_OUT in 1h 30m 53s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/74/662974/1/check/puppet-openstack-beaker-centos-7/f280f03/ : SUCCESS in 16m 26s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/74/662974/1/check/puppet-openstack-beaker-ubuntu-bionic/58eb365/ : SUCCESS in 12m 11s\n- build-openstack-releasenotes http://logs.openstack.org/74/662974/1/check/build-openstack-releasenotes/fdb0a69/html/ : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"43c731353bd35c491d956f474316cd1f4f7aabfc","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-06-04 14:40:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a6843cfb830bb3bb632f0dab818ec3c8eace647b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 16:27:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/74/662974/1/check/puppet-openstack-lint/4620c23/ : SUCCESS in 4m 20s\n- puppet-openstack-syntax-4 http://logs.openstack.org/74/662974/1/check/puppet-openstack-syntax-4/90aff8b/ : SUCCESS in 4m 15s\n- puppet-openstack-syntax-5 http://logs.openstack.org/74/662974/1/check/puppet-openstack-syntax-5/de24364/ : SUCCESS in 4m 35s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/74/662974/1/check/puppet-openstack-unit-4.8-centos-7/7b3bb58/ : SUCCESS in 10m 09s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/74/662974/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/2a17334/ : SUCCESS in 37m 51s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/74/662974/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/31969a5/ : SUCCESS in 48m 24s\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous http://logs.openstack.org/74/662974/1/check/puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous/c025c63/ : TIMED_OUT in 1h 30m 48s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/74/662974/1/check/puppet-openstack-beaker-centos-7/5110191/ : SUCCESS in 17m 14s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/74/662974/1/check/puppet-openstack-beaker-ubuntu-bionic/04331ba/ : SUCCESS in 13m 08s\n- build-openstack-releasenotes http://logs.openstack.org/74/662974/1/check/build-openstack-releasenotes/200dd1a/html/ : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d1c9cb21f4d66c039964f3f921e6b8b3189099e6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-04 17:16:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"50a5ed1a1f847c12ecba9c8533b21c04f121173d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-05 01:09:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a3f8772c06a64cfa68c7b67e08d9690486c62a85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 01:09:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"008014266c537d63c3445116231a09682450dc21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 02:05:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint http://logs.openstack.org/74/662974/1/gate/puppet-openstack-lint/49ed266/ : SUCCESS in 4m 07s\n- puppet-openstack-syntax-4 http://logs.openstack.org/74/662974/1/gate/puppet-openstack-syntax-4/c06a073/ : SUCCESS in 3m 27s\n- puppet-openstack-syntax-5 http://logs.openstack.org/74/662974/1/gate/puppet-openstack-syntax-5/12188ac/ : SUCCESS in 3m 49s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/74/662974/1/gate/puppet-openstack-unit-4.8-centos-7/38f0033/ : SUCCESS in 15m 45s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/74/662974/1/gate/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/f6547e5/ : SUCCESS in 34m 23s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic http://logs.openstack.org/74/662974/1/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic/d7ae60d/ : SUCCESS in 46m 44s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/74/662974/1/gate/puppet-openstack-beaker-centos-7/b4bd2b5/ : SUCCESS in 16m 36s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/74/662974/1/gate/puppet-openstack-beaker-ubuntu-bionic/92473ee/ : SUCCESS in 13m 49s\n- build-openstack-releasenotes http://logs.openstack.org/74/662974/1/gate/build-openstack-releasenotes/31aee81/html/ : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1},{"id":"694c5e5f8a2b69545414e1170369b2fcfd68fa6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 02:05:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0732d7b1162643781629b6122c1f2faeadc2de9","revisions":{"b0732d7b1162643781629b6122c1f2faeadc2de9":{"kind":"REWORK","_number":1,"created":"2019-06-04 08:45:23.000000000","uploader":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"ref":"refs/changes/74/662974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/74/662974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/74/662974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/74/662974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/74/662974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/74/662974/1"}}},"commit":{"parents":[{"commit":"f969ee6055cd7abf3072b13fc9244cc299329bb2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f969ee6055cd7abf3072b13fc9244cc299329bb2"}]}],"author":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2019-05-24 12:41:12.000000000","tz":120},"committer":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2019-06-04 08:45:23.000000000","tz":0},"subject":"Add parameters necessary to use a remote Ganesha service","message":"Add parameters necessary to use a remote Ganesha service\n\nThe CephFS backend can use SSH to configure Ganesha even when this\nis deployed on remote servers. This change adds the parameters\nnecessary to configure Manila with the appropriate credentials to\nconnect via SSH to the remote server.\n\nChange-Id: I31fff6a1b6fede860b8aaee0079183924e8b31d6\n(cherry picked from commit 80a8df65930fee17ed5c736e966f06242ff7e065)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/b0732d7b1162643781629b6122c1f2faeadc2de9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/b0732d7b1162643781629b6122c1f2faeadc2de9"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
