)]}'
{"id":"openstack%2Fpuppet-manila~672147","triplet_id":"openstack%2Fpuppet-manila~master~Ia5d56f3761e89e16e8a4c586502b2b0c31291dd8","project":"openstack/puppet-manila","branch":"master","hashtags":[],"change_id":"Ia5d56f3761e89e16e8a4c586502b2b0c31291dd8","subject":"remove releasenotes mitaka references","status":"ABANDONED","created":"2019-07-22 19:21:55.000000000","updated":"2019-07-22 20:25:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2dd7a70fbbe453b98afd06f14a599e709ecba7bc","_number":672147,"virtual_id_number":672147,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-22 19:34:01.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2019-07-22 19:31:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 19:31:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-22 19:34:01.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"e38ddbe7b4a8ae89301998a6e4721d49d5cc2b23","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-22 19:21:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd6e5836abad9b590c56d23ca81aefa71b7f4683","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 19:31:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes http://logs.openstack.org/47/672147/1/check/build-openstack-releasenotes/87dea27/html/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8efada86a3cd83c14c814c83f8223b3f6c8a1f0d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-07-22 19:33:22.000000000","message":"Patch Set 1:\n\nNo failures occur for me; EOL branches are deleted; and the reno tooling should not fail on this... \n\nSee: https://docs.openstack.org/reno/latest/releasenotes/index.html#newton-and-earlier-1-9-0-origin-stable-newton","accounts_in_message":[],"_revision_number":1},{"id":"8609176c24e71ec58e99de6a444e66270043dea2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-07-22 19:34:01.000000000","message":"Patch Set 1:\n\n$ git clone https://github.com/openstack/puppet-manila\nCloning into \u0027puppet-manila\u0027...\ncd puppremote: Enumerating objects: 128, done.\nremote: Counting objects: 100% (128/128), done.\nremote: Compressing objects: 100% (117/117), done.\nremote: Total 3813 (delta 62), reused 75 (delta 9), pack-reused 3685\nReceiving objects: 100% (3813/3813), 756.36 KiB | 5.40 MiB/s, done.\nResolving deltas: 100% (2484/2484), done.\n\n  $ cd puppet-manila\n  ~/puppet-manila$ tox -ereno\nreno create: /opt/stack/puppet-manila/.tox/reno\nsreno installed: DEPRECATION: Python 2.7 will reach the end of its life on January 1st, 2020. Please upgrade your Python as Python 2.7 won\u0027t be maintained after that date. A future version of pip will drop support for Python 2.7.\nreno run-test-pre: PYTHONHASHSEED\u003d\u00273973847338\u0027\n___________________________________________________________________________________________________ summary ____________________________________________________________________________________________________\n  reno: commands succeeded\n  congratulations :)","accounts_in_message":[],"_revision_number":1},{"id":"fcb3048a3c0ea7eb2cd82d8c33a88f1a30a06d17","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-22 20:24:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"2dd7a70fbbe453b98afd06f14a599e709ecba7bc","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-22 20:25:18.000000000","message":"Patch Set 1:\n\nI recloned puppet-manila and no longer have the issue.\n\nThanks, Goutham!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"076a21580ad891e268856e6a14bb45e25092b94b","revisions":{"076a21580ad891e268856e6a14bb45e25092b94b":{"kind":"REWORK","_number":1,"created":"2019-07-22 19:21:55.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/47/672147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/47/672147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/47/672147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/47/672147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/47/672147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/47/672147/1"}}},"commit":{"parents":[{"commit":"009b9c77c20b87a1e7e51ec83b39b1c23a474839","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/009b9c77c20b87a1e7e51ec83b39b1c23a474839"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-22 19:18:00.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-22 19:21:07.000000000","tz":-240},"subject":"remove releasenotes mitaka references","message":"remove releasenotes mitaka references\n\nThey cause ``tox`` to fail when run locally  because there is no\nstable/mitaka branch for this component.\n\nChange-Id: Ia5d56f3761e89e16e8a4c586502b2b0c31291dd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/076a21580ad891e268856e6a14bb45e25092b94b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/076a21580ad891e268856e6a14bb45e25092b94b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
