)]}'
{"id":"openstack%2Fpuppet-manila~672148","triplet_id":"openstack%2Fpuppet-manila~master~I99fe50cb5e112bcad85494dc755725d3444f717b","project":"openstack/puppet-manila","branch":"master","topic":"CephFS-NFS-IPv6","hashtags":[],"change_id":"I99fe50cb5e112bcad85494dc755725d3444f717b","subject":"Support cephfs_ganesha_export_ips parameter","status":"MERGED","created":"2019-07-22 19:22:26.000000000","updated":"2019-07-31 20:55:53.000000000","submitted":"2019-07-31 20:55:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672148-1564606553455-05bf691b","meta_rev_id":"47e01841d7bdff48c98045dca207f294e55d98a5","_number":672148,"virtual_id_number":672148,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":2,"date":"2019-07-31 20:55:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-29 17:15:11.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-29 17:54:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2019-07-31 20:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2019-07-31 20:06:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-07-25 20:42:10.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-29 17:15:11.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-29 17:54:21.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-07-31 20:06:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-31 20:55:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff0218fe76e33253ef1a538939b24805f94d7311","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-22 19:22:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee85cbb4320f463719ca66694344841f5076b37d","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-22 19:22:59.000000000","message":"Patch Set 1: Workflow-1\n\nWIP until the manila change merges.","accounts_in_message":[],"_revision_number":1},{"id":"2eb26122bf1972e4d18885a73d3badbab0a93c84","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-07-22 19:23:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7225cd0331d5ebea64a3fb2c7ac7eddcd210f271","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 20:10:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/48/672148/1/check/puppet-openstack-lint-ubuntu-bionic/852c624/ : SUCCESS in 4m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/48/672148/1/check/puppet-openstack-syntax-5-ubuntu-bionic/124a25a/ : SUCCESS in 5m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/48/672148/1/check/puppet-openstack-syntax-6-ubuntu-bionic/b0266be/ : SUCCESS in 5m 05s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/48/672148/1/check/puppet-openstack-unit-5.5-centos-7/18b259c/ : FAILURE in 12m 42s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/48/672148/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/ce9d7ae/ : FAILURE in 10m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/48/672148/1/check/puppet-openstack-unit-latest-ubuntu-bionic/dd460df/ : FAILURE in 9m 18s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/48/672148/1/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/55e182d/ : FAILURE in 25m 42s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/48/672148/1/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/91a914b/ : SUCCESS in 42m 53s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/48/672148/1/check/puppet-openstack-beaker-centos-7/44f47f4/ : SUCCESS in 19m 01s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/48/672148/1/check/puppet-openstack-beaker-ubuntu-bionic/4471f7b/ : FAILURE in 9m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e9ecf45d601d5424104f7fc128221d5182a0d2f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-22 20:26:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c94843008403b3acbfa9a343b58d8a1be107f911","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-22 20:29:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f7184a2d3b67d4141a5ab51a830aa8b38fa18bfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 21:19:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/48/672148/2/check/puppet-openstack-lint-ubuntu-bionic/36de493/ : SUCCESS in 4m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/48/672148/2/check/puppet-openstack-syntax-5-ubuntu-bionic/82f9fb7/ : SUCCESS in 5m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/48/672148/2/check/puppet-openstack-syntax-6-ubuntu-bionic/2f6f7a6/ : SUCCESS in 4m 35s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/48/672148/2/check/puppet-openstack-unit-5.5-centos-7/9e6ff7b/ : FAILURE in 11m 27s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/48/672148/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/fe3e790/ : FAILURE in 12m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/48/672148/2/check/puppet-openstack-unit-latest-ubuntu-bionic/8eaa944/ : FAILURE in 8m 08s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/48/672148/2/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/b14a413/ : FAILURE in 26m 37s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/48/672148/2/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/0734f49/ : SUCCESS in 44m 57s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/48/672148/2/check/puppet-openstack-beaker-centos-7/fa20633/ : SUCCESS in 18m 04s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/48/672148/2/check/puppet-openstack-beaker-ubuntu-bionic/329409a/ : FAILURE in 11m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b5c44d4d8aac27b029a63ba73d842e40fca1b4a6","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-23 11:27:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"64f3315c624df8648c2713bfb19a503f0e976912","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 12:17:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/48/672148/3/check/puppet-openstack-lint-ubuntu-bionic/199453c/ : SUCCESS in 5m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/48/672148/3/check/puppet-openstack-syntax-5-ubuntu-bionic/0baac0f/ : SUCCESS in 5m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/48/672148/3/check/puppet-openstack-syntax-6-ubuntu-bionic/a04a52a/ : SUCCESS in 4m 43s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/48/672148/3/check/puppet-openstack-unit-5.5-centos-7/098e222/ : FAILURE in 21m 09s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/48/672148/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/6eddc91/ : FAILURE in 12m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/48/672148/3/check/puppet-openstack-unit-latest-ubuntu-bionic/e6e4294/ : FAILURE in 15m 14s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/48/672148/3/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/607a3cd/ : FAILURE in 29m 14s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/48/672148/3/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/049c0b2/ : SUCCESS in 43m 11s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/48/672148/3/check/puppet-openstack-beaker-centos-7/f483071/ : SUCCESS in 18m 49s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/48/672148/3/check/puppet-openstack-beaker-ubuntu-bionic/7ee58f1/ : FAILURE in 8m 00s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/48/672148/3/check/build-openstack-releasenotes/0621ef5/html/ : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":3},{"id":"5ba4467d38ac40ba2da2f37c0a7661aec4ab94af","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-23 14:14:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c2b91f4e4d4253911fd75382a8160f9805142c5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:06:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/48/672148/4/check/puppet-openstack-lint-ubuntu-bionic/0554f3f/ : SUCCESS in 5m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/48/672148/4/check/puppet-openstack-syntax-5-ubuntu-bionic/33e6d44/ : SUCCESS in 5m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/48/672148/4/check/puppet-openstack-syntax-6-ubuntu-bionic/2ac802d/ : SUCCESS in 5m 26s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/48/672148/4/check/puppet-openstack-unit-5.5-centos-7/3978934/ : SUCCESS in 25m 01s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/48/672148/4/check/puppet-openstack-unit-6.4-ubuntu-bionic/2f4baf9/ : SUCCESS in 13m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/48/672148/4/check/puppet-openstack-unit-latest-ubuntu-bionic/a31f481/ : SUCCESS in 11m 51s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic http://logs.openstack.org/48/672148/4/check/puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic/b32db26/ : FAILURE in 27m 50s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/48/672148/4/check/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/5d96551/ : SUCCESS in 46m 22s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/48/672148/4/check/puppet-openstack-beaker-centos-7/90ef472/ : SUCCESS in 17m 10s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/48/672148/4/check/puppet-openstack-beaker-ubuntu-bionic/ced1618/ : FAILURE in 11m 11s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/48/672148/4/check/build-openstack-releasenotes/f93972c/html/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":4},{"id":"f958c3d517bf0e3ab06929abdeb41902183e49e6","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-23 15:14:43.000000000","message":"Patch Set 4: Workflow-1\n\nwait for the corresponding Manila patch to merge.","accounts_in_message":[],"_revision_number":4},{"id":"ba85159221c618166cb157dfcd19dafabbd77e59","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-25 10:47:35.000000000","message":"Topic set to CephFS-NFS-IPv6","accounts_in_message":[],"_revision_number":4},{"id":"6a9815aed1c32dfd0a8417f504981453f94a6872","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-25 20:42:10.000000000","message":"Patch Set 4: -Workflow\n\nmanila review 670510 has merged -- re-establishing workflow","accounts_in_message":[],"_revision_number":4},{"id":"1b40acdce9b72d2426485765e7e5bbc01a09670f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-29 17:15:11.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"05be84ca92f9620dd99ba06add0252688dbd469d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-07-29 17:54:21.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"68ee859aa13b1f71fa2f5b9e55d07979e7b5e52c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-31 20:06:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f1c43d916f384ec85c8b3fbfbca40bd4fcedfd45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 20:06:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1966af5e6488cbf0682de6069c707e18ff800f91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 20:55:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/48/672148/4/gate/puppet-openstack-lint-ubuntu-bionic/e2f6fea/ : SUCCESS in 5m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/48/672148/4/gate/puppet-openstack-syntax-5-ubuntu-bionic/bd1fa82/ : SUCCESS in 5m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/48/672148/4/gate/puppet-openstack-syntax-6-ubuntu-bionic/c759591/ : SUCCESS in 6m 45s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/48/672148/4/gate/puppet-openstack-unit-5.5-centos-7/849bd3e/ : SUCCESS in 20m 01s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/48/672148/4/gate/puppet-openstack-unit-6.4-ubuntu-bionic/a38d1fb/ : SUCCESS in 14m 39s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus http://logs.openstack.org/48/672148/4/gate/puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus/0a4aea3/ : SUCCESS in 45m 04s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/48/672148/4/gate/puppet-openstack-beaker-centos-7/ccfa7b5/ : SUCCESS in 20m 53s\n- build-openstack-releasenotes http://logs.openstack.org/48/672148/4/gate/build-openstack-releasenotes/3e21e97/html/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":4},{"id":"47e01841d7bdff48c98045dca207f294e55d98a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 20:55:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"97e1dab0dd15f3be02b12008019c94a6b4ee9312","revisions":{"0a8000a52cd31673d7b4dd25c8a22e3a2a71d967":{"kind":"REWORK","_number":1,"created":"2019-07-22 19:22:26.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/48/672148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/48/672148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/1"}}},"commit":{"parents":[{"commit":"076a21580ad891e268856e6a14bb45e25092b94b","subject":"remove releasenotes mitaka references","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/076a21580ad891e268856e6a14bb45e25092b94b"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-22 18:57:12.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-22 19:22:05.000000000","tz":-240},"subject":"Support cephfs_ganesha_export_ips parameter","message":"Support cephfs_ganesha_export_ips parameter\n\nManila now [1] supports configuring a list of IP addresses\nor hostnames on which NFS-ganesha will listen for mounts and\nprovide NFS service.\n\nExpose this parameter in the backend manifests for manila with\nCephFS.\n\n[1] https://review.opendev.org/#/c/670510\nChange-Id: I99fe50cb5e112bcad85494dc755725d3444f717b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/0a8000a52cd31673d7b4dd25c8a22e3a2a71d967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/0a8000a52cd31673d7b4dd25c8a22e3a2a71d967"}]},"branch":"refs/heads/master"},"b90b449d65c999f494866014c18dd5ad1b948a9d":{"kind":"REWORK","_number":2,"created":"2019-07-22 20:29:12.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/48/672148/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/48/672148/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/2"}}},"commit":{"parents":[{"commit":"9fb1805e1b16e44bfa9402eaaf22aba2a4449bde","subject":"Manage oslo_middleware/max_request_body_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9fb1805e1b16e44bfa9402eaaf22aba2a4449bde"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-22 18:57:12.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-22 20:29:12.000000000","tz":-240},"subject":"Support cephfs_ganesha_export_ips parameter","message":"Support cephfs_ganesha_export_ips parameter\n\nManila now [1] supports configuring a list of IP addresses\nor hostnames on which NFS-ganesha will listen for mounts and\nprovide NFS service.\n\nExpose this parameter in the backend manifests for manila with\nCephFS.\n\n[1] https://review.opendev.org/#/c/670510\nChange-Id: I99fe50cb5e112bcad85494dc755725d3444f717b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/b90b449d65c999f494866014c18dd5ad1b948a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/b90b449d65c999f494866014c18dd5ad1b948a9d"}]},"branch":"refs/heads/master"},"2b514251ce1fb194b735ae592364f0f57272e21b":{"kind":"REWORK","_number":3,"created":"2019-07-23 11:27:27.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/48/672148/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/48/672148/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/3"}}},"commit":{"parents":[{"commit":"9fb1805e1b16e44bfa9402eaaf22aba2a4449bde","subject":"Manage oslo_middleware/max_request_body_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9fb1805e1b16e44bfa9402eaaf22aba2a4449bde"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-22 18:57:12.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-23 11:27:14.000000000","tz":-240},"subject":"Support cephfs_ganesha_export_ips parameter","message":"Support cephfs_ganesha_export_ips parameter\n\nManila now [1] supports configuring a list of IP addresses\nor hostnames on which NFS-ganesha will listen for mounts and\nprovide NFS service.\n\nExpose this parameter in the backend manifests for manila with\nCephFS.\n\n[1] https://review.opendev.org/#/c/670510\nChange-Id: I99fe50cb5e112bcad85494dc755725d3444f717b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/2b514251ce1fb194b735ae592364f0f57272e21b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/2b514251ce1fb194b735ae592364f0f57272e21b"}]},"branch":"refs/heads/master"},"97e1dab0dd15f3be02b12008019c94a6b4ee9312":{"kind":"REWORK","_number":4,"created":"2019-07-23 14:14:41.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/48/672148/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/48/672148/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/48/672148/4"}}},"commit":{"parents":[{"commit":"9fb1805e1b16e44bfa9402eaaf22aba2a4449bde","subject":"Manage oslo_middleware/max_request_body_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9fb1805e1b16e44bfa9402eaaf22aba2a4449bde"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-22 18:57:12.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-23 14:14:29.000000000","tz":-240},"subject":"Support cephfs_ganesha_export_ips parameter","message":"Support cephfs_ganesha_export_ips parameter\n\nManila now [1] supports configuring a list of IP addresses\nor hostnames on which NFS-ganesha will listen for mounts and\nprovide NFS service.\n\nExpose this parameter in the backend manifests for manila with\nCephFS.\n\n[1] https://review.opendev.org/#/c/670510\nChange-Id: I99fe50cb5e112bcad85494dc755725d3444f717b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/97e1dab0dd15f3be02b12008019c94a6b4ee9312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/97e1dab0dd15f3be02b12008019c94a6b4ee9312"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
