)]}'
{"id":"openstack%2Fpuppet-manila~676546","triplet_id":"openstack%2Fpuppet-manila~master~I0cbf36194df9f94c173167557b5e57b57bf68a65","project":"openstack/puppet-manila","branch":"master","topic":"remove_deprecate_pki_related_options","hashtags":[],"change_id":"I0cbf36194df9f94c173167557b5e57b57bf68a65","subject":"Remove deprecated pki related options","status":"MERGED","created":"2019-08-15 03:52:30.000000000","updated":"2019-08-16 09:09:06.000000000","submitted":"2019-08-16 09:09:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676546-1565946546614-3f7c88c0","meta_rev_id":"3791eed54203b61898645bac4e94ad16d37a7986","_number":676546,"virtual_id_number":676546,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-08-16 09:09:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-08-15 16:23:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-16 08:14:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-08-16 08:14:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 16:23:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-16 08:14:15.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-08-16 09:09:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1e6f75c60abb94a1a8f235422558b0c733fc542","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-15 03:52:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57fed4551cc0da2db82030e1a2df2df4c6c35cc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 04:46:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c59c41b9bd6e41bd81101b157dff930d : SUCCESS in 6m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2308477895c46c48617285657e443de : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1838d4e22eb4bf5a4991288a4776d4b : SUCCESS in 4m 26s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/cc987a00adf84195862098c09239701c : SUCCESS in 20m 06s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4aa56a1e2d24677a663a9ea8b2c7264 : SUCCESS in 14m 30s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/728cb30950884d44953f9193cb79c3b6 : SUCCESS in 12m 53s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/80fb2a8677fd41ccad8fdde733e1196a : FAILURE in 23m 26s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/f7970545a2e2451b9e72fbb403fef44f : SUCCESS in 46m 37s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/018dc3af00f449d58a428b9c933e11d5 : SUCCESS in 20m 09s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1b98af13d0640598486bb11eaee81f6 : FAILURE in 9m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a822ac479324ec0aab1bc2c45cf6624 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9e4646ef1f627d7d7ceacdfccd5f0c53ba073ce0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-15 16:23:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6af8507496bb290c349f452e396d8e97bb753b4a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-16 08:14:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"91ea0f454a7505ac8c412e02f4d85d0f3ca1d5b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 08:14:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09217a65b6396453c8e83a71a547d93f02aeecec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:09:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"23601b741979705019b039627e8f3f480d9b1045","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:09:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64a550813c054874b7e70381e27b1102 : SUCCESS in 5m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f52a5b36b244d22be2b88521f87c34b : SUCCESS in 4m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84bd207556674891a0ad65cd79b4c2dd : SUCCESS in 3m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4d84fbca1c9a47148bef4017ab6cf28e : SUCCESS in 17m 26s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf9db1ae7c6344c6bb9abab5f1856fc5 : SUCCESS in 11m 32s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/f6e1ed214807404a91bdcd15625b32d7 : SUCCESS in 42m 25s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/cb684e2a82de4f18b546f6ef051a5c57 : SUCCESS in 20m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24365b0f49df494e9698c97f76b2463d : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f6f2d4d45522c685680d3b52c6eb7417c23de4a","revisions":{"1f6f2d4d45522c685680d3b52c6eb7417c23de4a":{"kind":"REWORK","_number":1,"created":"2019-08-15 03:52:30.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/46/676546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/46/676546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/46/676546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/46/676546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/46/676546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/46/676546/1"}}},"commit":{"parents":[{"commit":"3da0ff997f1107387343bd3962390a3ab0f3a37a","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/3da0ff997f1107387343bd3962390a3ab0f3a37a"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-15 03:51:37.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-15 03:51:37.000000000","tz":480},"subject":"Remove deprecated pki related options","message":"Remove deprecated pki related options\n\nThe deprecated pki related options check_revocations_for_cached and\nhash_algorithms option has been removed.\n\nChange-Id: I0cbf36194df9f94c173167557b5e57b57bf68a65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1f6f2d4d45522c685680d3b52c6eb7417c23de4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1f6f2d4d45522c685680d3b52c6eb7417c23de4a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
