)]}'
{"id":"openstack%2Fpuppet-manila~677633","triplet_id":"openstack%2Fpuppet-manila~master~Ibed69e1116543709fd0f51778b93c86bed571dcd","project":"openstack/puppet-manila","branch":"master","topic":"bug/1840868","hashtags":[],"change_id":"Ibed69e1116543709fd0f51778b93c86bed571dcd","subject":"Introduce the new rabbit_heartbeat_in_pthread option","status":"MERGED","created":"2019-08-21 06:25:20.000000000","updated":"2019-08-22 03:13:29.000000000","submitted":"2019-08-22 03:13:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677633-1566443609681-66c16398","meta_rev_id":"9e6213fc432d9368457f3a93dddfbd62d98bc8bd","_number":677633,"virtual_id_number":677633,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-08-22 03:13:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-22 02:24:40.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-21 07:33:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-22 02:24:40.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 07:33:54.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-08-22 02:24:40.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-22 03:13:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3fb57dc33061183c98c9bf26fece1f8126654fdf","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-21 06:25:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f5736ea03833a7a9065c560cb7bbdb077f36389","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 07:18:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69b109333c3b48d4afdefbd1aa02c3ea : SUCCESS in 7m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db03fa166eb348558f2cf8b9d24c3103 : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47acffa690d249b4b03eb54c83628a8a : SUCCESS in 4m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b040327475064b0f835a3ad33fb48e80 : SUCCESS in 20m 36s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0329b8de227c45d5ae5907b53c3ac602 : SUCCESS in 14m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d197d866b82402e8c858a872ba94063 : SUCCESS in 16m 38s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ffd6578438a64673a47ac2b108bc6d0b : FAILURE in 28m 11s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/b79d772eaa7245cd8af40f81ee7d4e2a : SUCCESS in 45m 30s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/185c95eb046543ff89ccfea0d6f2d54a : SUCCESS in 19m 00s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71af26db01dc46a091dc75753f712c87 : FAILURE in 11m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74cc7c04e78546e0a984ab27a73b18c8 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6e13c58a772b4bba0597a85c2b2eef74eb67b398","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-21 07:33:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"90d15e3ed383f7c5df080df37292c6a60812ff10","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 02:24:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1543b4bdf0940ebf367d8e41eea272006ff31f6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 02:24:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26dac7d20bd0af2739a69c1b249da02385438100","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 03:13:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"58d571a225cb6a0e08177ac837407a765ba230f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 03:13:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/304219243c8a46f98d92820732e5365c : SUCCESS in 5m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afe0a6fd95e14f2686ca615110d495c1 : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7090f2964104fa2b82c5320262286fd : SUCCESS in 5m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/20b905416c494652bcb406c8d72896da : SUCCESS in 20m 24s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25b4083d9f444d1d9d9207b6af24c2d1 : SUCCESS in 13m 39s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/41fee8ba0b2a47bca97b79a3686bc0fc : SUCCESS in 42m 50s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/eb726ccd17af417c92a41861197193d9 : SUCCESS in 18m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad90d430880440249be8a2e02eda3341 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d69a4f20a796a51e42ac5ab679ab3bd8a42458f2","revisions":{"d69a4f20a796a51e42ac5ab679ab3bd8a42458f2":{"kind":"REWORK","_number":1,"created":"2019-08-21 06:25:20.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/33/677633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/33/677633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/33/677633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/33/677633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/33/677633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/33/677633/1"}}},"commit":{"parents":[{"commit":"1f6f2d4d45522c685680d3b52c6eb7417c23de4a","subject":"Remove deprecated pki related options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1f6f2d4d45522c685680d3b52c6eb7417c23de4a"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-21 06:21:01.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-21 06:23:55.000000000","tz":480},"subject":"Introduce the new rabbit_heartbeat_in_pthread option","message":"Introduce the new rabbit_heartbeat_in_pthread option\n\noslo.messaging RabbitMQ driver have now a new option that allow user to\nrun the RabbitMQ heartbeat over a native python thread.\n\nThese change allow user to use this new option.\n\nChange-Id: Ibed69e1116543709fd0f51778b93c86bed571dcd\nCloses-Bug: #1840868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/d69a4f20a796a51e42ac5ab679ab3bd8a42458f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/d69a4f20a796a51e42ac5ab679ab3bd8a42458f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
