)]}'
{"id":"openstack%2Fpuppet-manila~683749","triplet_id":"openstack%2Fpuppet-manila~master~Iaeddc872771da78a71c6fc258aae40d5346becab","project":"openstack/puppet-manila","branch":"master","topic":"service_token_roles","hashtags":[],"change_id":"Iaeddc872771da78a71c6fc258aae40d5346becab","subject":"Add support to configure service_token_roles in authtoken middleware","status":"MERGED","created":"2019-09-21 01:29:37.000000000","updated":"2019-09-23 08:25:47.000000000","submitted":"2019-09-23 08:24:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683749-1569227081045-1ec4dea2","meta_rev_id":"fe43431bcf980840f341f4d0df9a6b5fa957d60f","_number":683749,"virtual_id_number":683749,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-09-23 08:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-09-23 06:20:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-09-23 07:33:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-09-23 07:33:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 06:20:58.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-09-23 07:33:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-09-23 08:24:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b53f4a2f1487a0603154952a5ff39cb0050ae876","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-09-21 01:29:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf7f4cab3c654b9abb1324ef642499f7ff59245d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 02:19:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/789eac9a31cc4cd498d8f879b8c4e1d6 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5560123155fd498995a0e2c8d846efda : SUCCESS in 4m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5c8ef7dbbd6423f92797ae7bc7c7b53 : SUCCESS in 4m 33s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/012718fa67084227a4462ed209fe4cb8 : SUCCESS in 19m 48s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a34fef29dc73415d8e7c0fc56fbcc3d1 : SUCCESS in 13m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90d42329a66e4a8cab1a019e3740882d : SUCCESS in 12m 12s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e721901aad6e43f6a62048ae537a0ded : FAILURE in 28m 55s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/8edb298be64549cebf834c7113d371a6 : SUCCESS in 43m 28s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d3505dd48d1a4f84b35694a379b04692 : SUCCESS in 15m 59s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4f5c82b81504e92ad954c64cfc6abf2 : FAILURE in 9m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f87e864117144d38a53d29d9e8e3426 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f6cde6e8b5b9780a37c6b511f0e0d600f5d33acc","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-23 06:20:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d3bd852b3dd370cc25b6dd88fe80b00a8c908bef","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-09-23 07:33:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8cf60455a0c79350b36d4c3d264460b29e38c703","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 07:33:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd4405de754ecd3577baa368ef8b9cd3df9b49b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 08:24:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/867ace578b134619b2f1c7dce4052db4 : SUCCESS in 4m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1255a21fa3494f4ba2a1ded8fc50c7b6 : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/785f8d710b5e431e8182545a952c1329 : SUCCESS in 4m 37s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/881ab2d12fd949b394da89a562503084 : SUCCESS in 24m 33s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a4a22e3ef8948cbabe29c9a93c1f63e : SUCCESS in 13m 35s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/387fc3575c3f4327b3b5e5b4a630a318 : SUCCESS in 45m 48s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d24e5ea9ad3d467f9d1c61352366bf48 : SUCCESS in 18m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c56c70a130664947bf4978135287c2fd : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e5732606731133b9c3d05f1082674614111b8dd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 08:24:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1f0a62a16feb92a5669ba69c343c5ae54e07f238","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 08:25:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1320418c1dc4404ca038c7affe7e02e5 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95d13797e5e44d3c6de9a17a987c134d9dd7379d","revisions":{"95d13797e5e44d3c6de9a17a987c134d9dd7379d":{"kind":"REWORK","_number":1,"created":"2019-09-21 01:29:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/683749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/49/683749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/49/683749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/49/683749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/49/683749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/49/683749/1"}}},"commit":{"parents":[{"commit":"c5f922c287bd1e2479d84b7a647c14480c4db20d","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/c5f922c287bd1e2479d84b7a647c14480c4db20d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-21 01:25:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-09-21 01:29:34.000000000","tz":540},"subject":"Add support to configure service_token_roles in authtoken middleware","message":"Add support to configure service_token_roles in authtoken middleware\n\nChange-Id: Iaeddc872771da78a71c6fc258aae40d5346becab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/95d13797e5e44d3c6de9a17a987c134d9dd7379d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/95d13797e5e44d3c6de9a17a987c134d9dd7379d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
