)]}'
{"id":"openstack%2Fpuppet-manila~696806","triplet_id":"openstack%2Fpuppet-manila~stable%2Frocky~I76482e6ffd40920903f8b4d7fe2f608d8c12bdcc","project":"openstack/puppet-manila","branch":"stable/rocky","topic":"bug/1854443-stable/rocky","hashtags":[],"change_id":"I76482e6ffd40920903f8b4d7fe2f608d8c12bdcc","subject":"Configure service_instance_password as secret","status":"MERGED","created":"2019-12-02 02:53:35.000000000","updated":"2019-12-03 18:20:07.000000000","submitted":"2019-12-03 18:20:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696806-1575397207632-aa01f648","meta_rev_id":"be2fc92cd7a5e8dcf38a61566f8684701317b1ea","_number":696806,"virtual_id_number":696806,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-12-03 03:56:56.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"date":"2019-12-03 15:46:43.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-03 18:20:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-12-02 13:10:13.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-12-02 08:03:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-12-02 13:10:13.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-02 08:03:38.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-12-02 13:10:13.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-12-03 15:46:43.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-12-03 18:20:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53c811c7da0829ca043947ac1577a30ebd33c157","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-12-02 02:53:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4278d75e11d542533a3095866fa363022c8ae837","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 04:26:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/93a08a7a30594a15861f47c99f08c063 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/5b098ef214ed4ebcac9a7009066f3a12 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/abefcfb010ac4435b8e3746fe9852da2 : SUCCESS in 4m 03s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/5413dfc1ce574c1faabf0f12b22d19c7 : SUCCESS in 10m 03s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/56b6763d7cc945d68b81e1790f4bf651 : TIMED_OUT in 1h 30m 48s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/733b484ba732433989bc5e9bf788b003 : SUCCESS in 41m 57s\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/1d785f19fd3f40928707685ba79a1f8f : TIMED_OUT in 1h 30m 47s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/438b4076c8e94213a725f6d1069e95d7 : SUCCESS in 16m 09s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc2d7dd97f70433583514ce6c31ca694 : SUCCESS in 12m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ae75ce1b2bbc1122a5ba3384627394ad1459257c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-12-02 08:03:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"754e30e905d06ff431bed999da81665e51654b62","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-12-02 13:10:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c45504cd36006646390512287ea3b02cbeb27a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 13:10:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c48028865b74d00f1be57d6a43f4b8df9f753725","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 13:59:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/63fb5f7c0b9a4a6fa89bf5b5bd374f87 : SUCCESS in 4m 43s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/324e4a30b2cb4d388632006cde9a3326 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/2f6f55f14fb246c79cd81d5c42246556 : SUCCESS in 4m 15s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/2bec146065a94a9d9bc47a7cd40aad71 : SUCCESS in 14m 59s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/6c4282334c7f4bcc986fd9b21083f354 : SUCCESS in 40m 13s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/57c59453de284f25aa7a559fc98f80d6 : RETRY_LIMIT in 5m 38s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09f414e39eae4527b87e047cf0425deb : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":1},{"id":"88d5bde1ccfc564b5f4a635a4ecae181ce5871e0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-12-03 01:24:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"42f132ae226f9cffac7f505bb12a51f0da630c8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 03:19:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/00ff86b25f29458d8c6fefa1ac886dbf : SUCCESS in 3m 57s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/b00d4a9f461e41e387f43510943aac72 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/df157f6cc32f498392af86b057789849 : SUCCESS in 3m 59s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/1e73ca449ef54eaf8e01b8fe6e433896 : FAILURE in 6m 23s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/bc2f9e1abff3438da461328d4272586a : TIMED_OUT in 1h 30m 48s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/d0cf40d48a1f44f2b48c812c85842c72 : SUCCESS in 44m 30s\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/3aea9337e6f64af2b64e45ae7dfa55d7 : TIMED_OUT in 1h 30m 55s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/16b0c34406484e62829eceaeb896b53f : FAILURE in 6m 33s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15b0f13b15a045e6bc461cfed89cb976 : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":1},{"id":"fa566db344a12ec7db499d04907b3b9780664010","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-12-03 03:56:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"82767cdc11ced5e246c5c0a317b569d958e65e5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 05:33:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/97cdad801cfa4ba0a35b01066dc29ba1 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/5cbeeda460c748ce931910ff7181ac0b : SUCCESS in 3m 37s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/19aaee7188df4fdbb70ff01f629771ef : SUCCESS in 3m 35s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/2f4c52960a7741869cee9c578bcb9dc9 : FAILURE in 6m 09s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/617cc4ff9b824837a9c1e60066fd55d3 : TIMED_OUT in 1h 30m 41s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/da795de90a124018894cb3beade512a3 : SUCCESS in 41m 40s\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/b11b4f9a2d59447bae63e50dda1997db : TIMED_OUT in 1h 30m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9d0ab0b5ec884ebe834c31c6292affd8 : FAILURE in 6m 39s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/713d3182e785464caf54c35d09a9aff6 : SUCCESS in 10m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c4cdd9686bb544b0e155d53b3050f8a99debc734","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-03 15:46:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"97b247c8ba2d0505bff14d01312eac51cdcd490b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 17:27:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/0097ff39a91b4733baeae46c73a8867e : SUCCESS in 4m 56s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/a1d76c33cb8c4c49a993538569dcf4ea : SUCCESS in 4m 49s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/c1169056af9a4ba98645fdd523dcc1c7 : SUCCESS in 4m 18s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/aa8d83103493451da34a83591a3e6431 : SUCCESS in 17m 34s\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/23192b9365bb4b0e9dfca417fe75eb9a : TIMED_OUT in 1h 31m 09s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/b4882ad39e5746238948f6bf09b0f61b : SUCCESS in 43m 37s\n- puppet-openstack-integration-4-scenario004-tempest-debian-stable-luminous https://zuul.opendev.org/t/openstack/build/9b6148f92163443ebde7a504c2ae46e4 : TIMED_OUT in 1h 31m 04s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1978e56f30d4423e87bf63f5ac4a1c47 : SUCCESS in 17m 09s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2ae96dbdd1849b68aa7a5cf6bc58a41 : SUCCESS in 17m 14s","accounts_in_message":[],"_revision_number":1},{"id":"abe1f69f7b51ae187556932416cdbd8a9deab587","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 17:27:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e2f8b25a09952b6e5af4b043f7e6a01faf852d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 18:20:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/5ccf6c6b50554ee7b51df0f547b04a2b : SUCCESS in 4m 29s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/8cdf391d90d949d289e53099276a182d : SUCCESS in 3m 54s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/ba557efa799a4ac287e133a7999edd25 : SUCCESS in 4m 54s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/0e0aabd2e8cb46b8b9646f3f382d2a9d : SUCCESS in 18m 47s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-mimic https://zuul.opendev.org/t/openstack/build/929fcf2a2d4e4a2583a4dfc858885dd6 : SUCCESS in 46m 49s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0880e84443554bd0be446c67283992f3 : SUCCESS in 18m 49s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c58b0219e4b347f2bf64cb70959cefde : SUCCESS in 15m 10s","accounts_in_message":[],"_revision_number":1},{"id":"eb546a68b6c34ea7a0b2a5b7982a58024fefc3d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 18:20:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b268d2028e2cd52cf63c3bbbf8e38f744fb8eed5","revisions":{"b268d2028e2cd52cf63c3bbbf8e38f744fb8eed5":{"kind":"REWORK","_number":1,"created":"2019-12-02 02:53:35.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/06/696806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/06/696806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/696806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/696806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/696806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/06/696806/1"}}},"commit":{"parents":[{"commit":"001ed3329c86d0f93bff0d6a0b1b32a8b2c638d1","subject":"Update module version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/001ed3329c86d0f93bff0d6a0b1b32a8b2c638d1"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-11-29 07:21:50.000000000","tz":480},"committer":{"name":"zhongshengping","email":"chdzsp@163.com","date":"2019-12-02 02:53:35.000000000","tz":0},"subject":"Configure service_instance_password as secret","message":"Configure service_instance_password as secret\n\nParameter service_instance_password should be configured as a secret.\n\nChange-Id: I76482e6ffd40920903f8b4d7fe2f608d8c12bdcc\nCloses-Bug: #1854443\n(cherry picked from commit 4a061e27d36f101870bc111a4d6b5796630adfd2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/b268d2028e2cd52cf63c3bbbf8e38f744fb8eed5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/b268d2028e2cd52cf63c3bbbf8e38f744fb8eed5"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
