)]}'
{"id":"openstack%2Fpuppet-manila~723232","triplet_id":"openstack%2Fpuppet-manila~master~Ibe99e25ddc1145d49229b5c2e32e016d63e9b476","project":"openstack/puppet-manila","branch":"master","topic":"cleanup","hashtags":[],"change_id":"Ibe99e25ddc1145d49229b5c2e32e016d63e9b476","subject":"Cleanup deprecated service_instance_network_helper_type","status":"MERGED","created":"2020-04-26 14:37:20.000000000","updated":"2020-04-27 15:47:18.000000000","submitted":"2020-04-27 15:44:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723232-1588002297636-0ef6f594","meta_rev_id":"f39649cc0fa9f6080079ee7e7421268b662bfec3","_number":723232,"virtual_id_number":723232,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 15:44:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2020-04-26 16:03:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-04-27 14:57:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-04-26 16:01:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-04-27 14:57:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-26 16:01:43.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2020-04-26 16:03:33.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-27 14:57:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-27 15:44:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc7d68af17d292359e79d11a7bb6ef3c4ac70177","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-26 14:37:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a87e0f23a1bcc130f22d699c187a972dbded75f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 15:28:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c623fc07a664ff49adf3c30c04f8067 : SUCCESS in 5m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6488b6035cbe42c49d5466969802693a : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8b70e40de1e41a48b804e90078e86d3 : SUCCESS in 5m 12s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/962e20c349b04a4287ca9d678b328405 : SUCCESS in 16m 15s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/dee1c34a18bb4e6689d844551edb0c46 : SUCCESS in 16m 45s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b2077812e5b49b790d53aa0b7b9cbe0 : SUCCESS in 13m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3a4a8fb8213412785bfb15043c9853a : SUCCESS in 14m 32s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/faa773217f354bb989d6ff20f1dbea47 : SUCCESS in 41m 57s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b7478e08f7c5426c829ee93e4b37bea9 : SUCCESS in 49m 11s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/15f4e4bba36443a5a6d3fee48dae0fd3 : SUCCESS in 23m 19s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/9a26d2aa0daf4d4cb33a645895113fce : FAILURE in 12m 42s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb1c0025180d46f2ae82a642c2ed91da : FAILURE in 6m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec25af7a28134860afbee78ea5c38806 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"659465f5e0ca9224626f16b910fecc9c0b0379c2","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-04-26 16:01:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"036fb8c5ed311504e6c76f98aead60ee4817b737","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-26 16:03:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"88145a326e1476832836d65999d71e3405996058","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-27 14:57:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a21db48a9bb9a2534f32e51f9146530f1825193f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 14:57:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c80b291ef358668219cf532a07a3b8b21a35154d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:44:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8dd683ec5156e7ed96a29ea2e26e485328c5e2d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:44:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ac488e3a90640e1a34c1ee81168119e : SUCCESS in 6m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7180296c91754d61b0de4044cfad2fbc : SUCCESS in 5m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a75a5b8a4dbe4bcb9d8b41194be2c1a7 : SUCCESS in 5m 28s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5da81784eb4c477682a8b0db888ae3a8 : SUCCESS in 17m 10s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/88c597aa8de94a4e9612c578e91f0ca5 : SUCCESS in 15m 14s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e258acc77914d78a2f5bd84af892809 : SUCCESS in 12m 07s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9b8022d82c26499d9e12e12bcc6e639b : SUCCESS in 44m 58s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/890dec5b7b5a4a7a83b9c64ca916441b : SUCCESS in 21m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/746e0e823a124f769355276f4c040ce0 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"7ab218431044d0f78110a8f7cf0d9965b7f304c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:47:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11d48f77a3db478b8b840424b0093343 : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5732af782b3962ad44530919549d7d58d456689e","revisions":{"5732af782b3962ad44530919549d7d58d456689e":{"kind":"REWORK","_number":1,"created":"2020-04-26 14:37:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/723232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/32/723232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/32/723232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/32/723232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/32/723232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/32/723232/1"}}},"commit":{"parents":[{"commit":"d73580062569b8bb6423c14830ae39dba63c204b","subject":"Prepare Ussuri M3 again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/d73580062569b8bb6423c14830ae39dba63c204b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-26 14:36:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-26 14:36:59.000000000","tz":540},"subject":"Cleanup deprecated service_instance_network_helper_type","message":"Cleanup deprecated service_instance_network_helper_type\n\n... because it was deprecated quite a long ago[1].\n\n[1] 5955b37b9b640e6ffdd7d4f4fa20677187b26ad7\n\nChange-Id: Ibe99e25ddc1145d49229b5c2e32e016d63e9b476\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/5732af782b3962ad44530919549d7d58d456689e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/5732af782b3962ad44530919549d7d58d456689e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
