)]}'
{"id":"openstack%2Fpuppet-manila~723979","triplet_id":"openstack%2Fpuppet-manila~stable%2Fussuri~I16239450a4d0bb6c99294a5100fa9883c1580751","project":"openstack/puppet-manila","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I16239450a4d0bb6c99294a5100fa9883c1580751","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-04-28 14:54:09.000000000","updated":"2020-04-29 02:48:13.000000000","submitted":"2020-04-29 02:46:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723979-1588128394239-047f6eff","meta_rev_id":"5fa0a1d21f37cf3013cec94d7c0f5050f4ec2776","_number":723979,"virtual_id_number":723979,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 02:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-29 01:49:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-04-28 16:46:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-29 01:49:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 16:46:08.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-29 01:49:10.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-04-29 02:46:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f6d6ae010cc49dfc6d5033f36720a107b20ca35","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-28 14:54:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ab1ac4fa9a7435e37159712c097b1d06f7e19a4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-28 16:46:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"94e2e7b407084097152f9080ff15cca53b361550","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 16:50:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee5aeb8c56b44f0ba5541dd8fa73464e : SUCCESS in 7m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10a49f646c5b4284a6be4c63e2ac11e2 : SUCCESS in 6m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5549d7d0e40744328f6c580a4827ce8f : SUCCESS in 6m 07s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/02d50dce26f9492a9db1c9407437560e : SUCCESS in 15m 43s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/1031b6bae6394a87ab65efcfdc8e8bc6 : SUCCESS in 21m 10s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e127cbfdc43d43ac88ee8bd10d76a09f : SUCCESS in 13m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84e7fd77b9d3421dab159e98cb199826 : SUCCESS in 15m 12s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/77a6343094ea4558b383aea85dfffa77 : SUCCESS in 39m 44s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8d0508f99cc74b6b915af6ee4a098352 : SUCCESS in 49m 24s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/33f928aa7db445e6ad4e97f4ad025bcb : SUCCESS in 23m 04s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/78ff0bca72974c5b8c67161b71aa72a0 : FAILURE in 13m 52s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78ade542dc7140b699822027ed98af64 : FAILURE in 10m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bce4307d0a14354acb19d95c0433057 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"96c38d44e87d95b0cca792c07b61adda453d02e0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-29 01:49:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"41ed831b7d8a9ac4b1fcd57bc87167038a678d20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 01:49:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d7d9cca1a67ab4f4504d2d161b9f879aa93f915","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 02:46:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f6fa6afb468488cbc42f85cf6eecf0d : SUCCESS in 5m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cea9ee49dce4ea0b6eaca7cac0eb37e : SUCCESS in 5m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef2d437021ed4c2ba1298a64efef190e : SUCCESS in 5m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c20f5fa607744e868f53906bf4a623ce : SUCCESS in 15m 44s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/99aaf2da2a504b718e9546cc2ad97d6f : SUCCESS in 13m 38s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d292b4110b945c294a2c3094591c986 : SUCCESS in 12m 04s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ec92ebb1c73b4641933d3b99e772c8fe : SUCCESS in 46m 58s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/70a5bbd1d3bd44658421cdff70f0b183 : SUCCESS in 18m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcaed61695e84df481f2c0d5199a3719 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4e14d44e8026e65ec46777fdbd31e613c070973b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 02:46:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"54710de4a94b65d799bd450a4403c67f706ebea8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 02:48:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71338f700c8e47feb7f310232b1e0dbd : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ec75411ab046e3b66dadd220a8386f342676729","revisions":{"0ec75411ab046e3b66dadd220a8386f342676729":{"kind":"REWORK","_number":1,"created":"2020-04-28 14:54:09.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/79/723979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/79/723979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/79/723979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/79/723979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/79/723979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/79/723979/1"}}},"commit":{"parents":[{"commit":"b3c4f3983992e5cc1f57af98bab39ddadf5697ca","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/b3c4f3983992e5cc1f57af98bab39ddadf5697ca"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:54:07.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:54:07.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I16239450a4d0bb6c99294a5100fa9883c1580751\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/0ec75411ab046e3b66dadd220a8386f342676729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/0ec75411ab046e3b66dadd220a8386f342676729"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
