)]}'
{"id":"openstack%2Fpuppet-manila~745606","triplet_id":"openstack%2Fpuppet-manila~master~I47c795016c10f3afb7df4697ecf78880b8063b5a","project":"openstack/puppet-manila","branch":"master","topic":"cleanup-deprecations","hashtags":[],"change_id":"I47c795016c10f3afb7df4697ecf78880b8063b5a","subject":"Fix deprecated glusterfs option","status":"MERGED","created":"2020-08-10 23:56:23.000000000","updated":"2020-08-13 00:42:45.000000000","submitted":"2020-08-13 00:41:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745606-1597279268400-480d4010","meta_rev_id":"2a49ea03fe7d6de7d7deb89bbeb0884109ac0f52","_number":745606,"virtual_id_number":745606,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-13 00:41:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-12 15:26:29.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2020-08-12 13:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-08-12 15:09:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-12 15:26:29.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-12 13:57:47.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-08-12 15:09:28.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-08-12 15:26:29.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-13 00:41:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19bdd14f2fb7b9fc6d6979806937b6a8d26413f0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-10 23:56:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0926315d11059305a0626e6234719380b195c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 00:55:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c4f084d57de4e3b9e6102f77897dcc2 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d265d13bf0e64cb9aa3f95991c851b10 : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5c04c2810b64682ae6fc25c351b3d4a : SUCCESS in 4m 48s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6ca74e55db3c4929b3b88da9b67812dd : FAILURE in 15m 34s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9fd3a1e215c84887bf1e8cc2f33fe85d : FAILURE in 13m 01s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c6074988985446e96e84cddfc785b2b : FAILURE in 12m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86490bc448064ddc9cda87cec936dc10 : FAILURE in 6m 18s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9bbd1cb9fbc04f3a8acd208799d15d0b : SUCCESS in 46m 43s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0b570bd364c745db8bb1662629452917 : SUCCESS in 56m 53s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/18349e42286a4c018f49eb73b9485bca : SUCCESS in 23m 36s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/834b906be36845d3a792ca07fba93022 : FAILURE in 13m 46s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31c0c8aa82a344629791e1e3f30f9305 : FAILURE in 7m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9638c2f305dbee7ea2205c5efd69278965ac2af3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-11 04:57:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3a0584a746fe92d7bd733170edd5e6c378ba2871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 05:59:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ace13503e07a4fa9b46018e45a49ddc2 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/076d7aed77ab40eab381466989e63590 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b9897a0a66647138acdf4b2889ba656 : SUCCESS in 3m 44s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/09e79ca43295409b825ba87343afb4b1 : FAILURE in 9m 39s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/44cd2bb2ff9b4d448d1fedd6d7585eeb : FAILURE in 10m 21s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6842db39e15a42cb8f60a3a9724077f4 : FAILURE in 8m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7660a3629044a26a8ad757edbda149a : FAILURE in 9m 16s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b1bd4e9ce00a475ebac963176e79d9d9 : SUCCESS in 39m 52s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c95dac110b5c4b009a4670633e6fe22b : SUCCESS in 55m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0a74a6c6ab43431e91b568e8f8c808f1 : SUCCESS in 15m 04s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/74654395611d4cef83b9e02b24d6be34 : FAILURE in 8m 22s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34be9b2618f24ad9a4648ede5506ee5a : FAILURE in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8e89c2ecc276837142fd7be2202870d8dca15bdf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-11 09:30:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e4ed4e2468b3adfdca9b8b16d8659f28f4baf14","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-11 19:46:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"298752e4fb98d86c433f2033af5f7f5a989904ed","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-11 19:48:41.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks for your review! Please see the latest patchset.","accounts_in_message":[],"_revision_number":2},{"id":"5dab9ad0602b69fa402efedce5570dae5e2efa98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 20:44:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e638e22538e400a8766566871be38b3 : SUCCESS in 4m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9935f92b450f4a2cb75a5fa2d0a7cced : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81150a196c3b4537a7373272dd5efb1f : SUCCESS in 3m 43s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3a0655c652c04961b46b10ae52369f0a : SUCCESS in 14m 24s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/21a9b24213414e73818cca10c62a1b23 : SUCCESS in 13m 41s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc9c68610819425585a16f62fc7197c2 : SUCCESS in 10m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30aafe797181403795a7667b8d6e563c : FAILURE in 6m 55s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e41226e00d554569a8582f7cd9b885ab : SUCCESS in 36m 05s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/834cb22d73194e279c69b27394dcb6e2 : SUCCESS in 56m 13s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4a159455d6eb43b39e9998687cdcf314 : SUCCESS in 18m 46s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/becb49ab34194089a0bb84a890bb4821 : FAILURE in 9m 02s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c76619aa59d40fb93cebbc9f9d10b17 : FAILURE in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92c98ac750e238d3996bde1a937d9580c7d8b458","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-08-11 21:02:17.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks, Goutham!","accounts_in_message":[],"_revision_number":3},{"id":"a50191ad5122a3bdfca2caffd81bd0b9c6024d91","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-11 23:03:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"54447d48276e98470e4204b66d2d3acc53b72b74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 23:45:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2dc3556721f4833aa126e82bfaa6d11 : SUCCESS in 5m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f257d906929c485a87f64186ea496420 : SUCCESS in 4m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0153ac4ace3a4d65af74260295aa02ca : SUCCESS in 3m 31s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/751116d6f7ea435eb1e192d1842a5aae : SUCCESS in 9m 20s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c4b77602813645c993053dd40c826ef8 : SUCCESS in 14m 38s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39d2c0290c1e475ea37d27b73c47ccad : SUCCESS in 13m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01a4f27137b946e3b9052866c559ac7b : FAILURE in 10m 38s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6b76ad96f3d44e60a9ec712cef452dc9 : SUCCESS in 39m 57s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/965c8b7ca4de432181580500fa4ba57d : SUCCESS in 38m 40s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6e8849f3acb54e23a6a65a703da7e65c : SUCCESS in 14m 45s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/0c697a9906c240b6b5277b2d7d85317e : FAILURE in 12m 34s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1363d0ad12a40b3bd4cf0bb7a641bad : FAILURE in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"05887ae221ad8d46da8e095cbc1f888e3ff5c633","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-11 23:51:29.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nI\u0027m afraid that you accidentally overwrote the patch. I don\u0027t care which patch we review in this url.\n\nCould you add a release note about deprecated/new parameters ?","accounts_in_message":[],"_revision_number":4},{"id":"df7939a9da8d94b026741b5ce9eb6191702dc121","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-12 00:38:25.000000000","message":"Patch Set 4:\n\n\u003e I\u0027m afraid that you accidentally overwrote the patch. I don\u0027t care which patch we review in this url.\n\nArgh :( I do this too often\n\n\u003e Could you add a release note about deprecated/new parameters ?\n\nSure, will do!","accounts_in_message":[],"_revision_number":4},{"id":"7b87e9201cabb517d4d28b98e7ce6112af15e8d5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-12 00:48:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"199928cfb980f1cbedb45fd7930bb07c2f5c41df","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-12 00:51:47.000000000","message":"Topic changed from cleanup-deprecations-patch3 to cleanup-deprecations","accounts_in_message":[],"_revision_number":5},{"id":"4bea83a131d2d69ae7af1c811a959e08e3b33404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 01:39:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9852f53ae46f4fef8c9c8e06a3361c28 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e52fc872346549ebb53c6129ce468256 : SUCCESS in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05956b5b4b6a4038b08b0ec84804699f : SUCCESS in 3m 28s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d7fb1c1539f546818379ccc8c0edd129 : SUCCESS in 11m 05s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f20c55eb88b74e0e859b40f02cecbd11 : SUCCESS in 11m 56s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67fdfaab74304c86b69d25ae076badaa : SUCCESS in 7m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba57d1a8cbbd4485a80579486f22087c : FAILURE in 7m 14s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4b666f27f3e44d6091d3ddd62bca4622 : SUCCESS in 39m 16s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/72ff54a70622497dbfda1b5ffa684bc5 : SUCCESS in 45m 27s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/260ced72560b4bf5abc82538f2d08327 : SUCCESS in 17m 33s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/17a2d728841f495fab279ce8d4dda7d7 : FAILURE in 9m 36s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27e438aa8e134f888d015b1e2da1f5dd : FAILURE in 6m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed2df442b8884dd2920e6ee3f60a2646 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":5},{"id":"c70abadcde876a6ee4d5f49db75e77922a5791f7","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-08-12 10:51:48.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f04ed8d1f92215bac72c62b127c56710ff533d08","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-12 13:56:14.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"c89b1a27ca327a5f2a2d2f7b1cc191d09b92758f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-12 13:57:47.000000000","message":"Patch Set 6: Code-Review+2\n\nThe change looked almost good to me.\n\nI fixed indent and updated parameter name to make future removal easier, and hope my change makes sense to you.","accounts_in_message":[],"_revision_number":6},{"id":"2727580d46680beac6c7543e86eb284a9bf5b8f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 15:03:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d7da89003804690b65d259e0728535b : SUCCESS in 5m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f7fc15144954adca7b0dabbf6afb01b : SUCCESS in 5m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76d994f857b14c90a13f945a2c769d03 : SUCCESS in 5m 09s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/547a03f5a1b147f7b1054fd4498a8c44 : SUCCESS in 13m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c9d5e6dc3bee46f0bb058196a1c213f9 : SUCCESS in 13m 52s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03c1c78fd4774b268c756a643a4828f4 : SUCCESS in 12m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2f49c3f2df34dfcb575307a401cf424 : FAILURE in 7m 36s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5a7365990d894790a0fcb3f52c67c9bd : SUCCESS in 41m 29s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9888cba478424a57ba54123ee98db3af : SUCCESS in 1h 00m 05s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/195c14f94a434e5c98f3e6ff89947882 : SUCCESS in 18m 21s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/28fdb65cc65041d6a2e18de5e390132e : FAILURE in 9m 52s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c9bcc46d512402c8ea70c98ba9603e2 : FAILURE in 8m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0450d3fbbbbf41b9bb042d7459b75c4c : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":6},{"id":"c2c05e0496a277ed42dc08c4940bf9616f324ae4","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-08-12 15:09:28.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"839bd3a6560cf022de5ad1ee2d52104be5e7951a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-12 15:26:29.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"9d952e672f3d51449451c4358b33ba6950fbbeb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 15:26:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"09e8034aedbb9c131f09f027184a6f32272d98c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 16:45:49.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16e9aaf86b2d408db70772bff4a3f7eb : SUCCESS in 5m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3449e5d2f2cd467694a51b043f35c8a2 : SUCCESS in 6m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86232dcf1f124eaf87a925af781b9f4c : SUCCESS in 6m 09s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f38543aeec664853b5704b703d31dd05 : SUCCESS in 21m 37s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f8ac905592034bf8bbd05de43b3a1c45 : SUCCESS in 20m 05s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f0dfd17af4f400cb2445b3a257abc3f : SUCCESS in 11m 41s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7a1e990c2cd24e9c9e845a74f34f3a2b : FAILURE in 1h 16m 55s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/36422557c9d34a999cb31f3085d6c831 : SUCCESS in 24m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/528f3a3fb60f489d9f848be6baad775c : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":6},{"id":"80c4a3e099873a0b86942ab09ac386e7f3ecbc17","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-12 17:34:47.000000000","message":"Patch Set 4:\n\nrecheck\n\n\u003e I fixed indent and updated parameter name to make future removal easier, and hope my change makes sense to you.\n\nNo problem at all :) Thank you for doing that!","accounts_in_message":[],"_revision_number":4},{"id":"fdb9b5a61daa9377a069e98d86173b89117c1197","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-12 22:52:53.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"38694293061515cbb0fb1f56c24c3736a7891e4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 23:48:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf38838d9d1d4e92a51c3327989cc9c5 : SUCCESS in 5m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d9b17cd5788491394f5458d24a9dbeb : SUCCESS in 4m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4dc22903e4564886baa9e1d7f7933551 : SUCCESS in 4m 48s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4497a533aa0445fab130cead6b0664d9 : SUCCESS in 13m 00s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/78dae6a2c0a84ad89d5cdcd3cb745ef5 : SUCCESS in 14m 12s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b613d4d30d146af97cab2396151370f : SUCCESS in 10m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77a834061c5742d7a491b04e6e82e743 : FAILURE in 8m 10s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/77d01cbb23364b81b4cde1c0d07cdc9b : SUCCESS in 38m 29s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c4a06dd126be45f5bd5b7279908c5a62 : SUCCESS in 53m 08s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7f39c6551c0541718d3717b226eb317d : SUCCESS in 18m 10s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/11e58c8d6bf64d829c8fff10aa7fac8b : FAILURE in 10m 15s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2a01b8b808a498b8f9a86a731cfbbea : FAILURE in 5m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07db06598d9c4ce8a8338978b9911bdd : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":6},{"id":"779d8a9fc615621fef8a3d13e2606f82e9e3caae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 23:48:50.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5a98855268a5b41c4ea48863a0f24a96da5116d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 00:41:08.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed4cff96d8714e6d840b2048833b2630 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b148dd333e140c6ac28c054b00f1a2c : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c6ddcbc4bd14c8fa4ccd0e0e1ca490d : SUCCESS in 4m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a404b0c6a0e7442b8c5904b3a62d8379 : SUCCESS in 14m 20s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/fbb02645fc0d42f49ed8021287e0c81a : SUCCESS in 11m 49s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d49c8cd5328543598a3959735e9deb69 : SUCCESS in 11m 04s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2adf32c6537945a38bfe22325b0bbccc : SUCCESS in 47m 32s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/22ca5e9172a345a7a09594e99a77e72a : SUCCESS in 20m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9a34031842646b1b1a60d0e0c02597d : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":6},{"id":"590e82b858f4a62520edd559fd79cc883620a966","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 00:41:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"2a49ea03fe7d6de7d7deb89bbeb0884109ac0f52","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 00:42:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35e6df58395e4a98a0aa6841a046a560 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bacfbc1c661a089c879fb66b1a965231254a9e0d","revisions":{"02163665486062aabb53023c464e4706259850f9":{"kind":"REWORK","_number":1,"created":"2020-08-10 23:56:23.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/06/745606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/06/745606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/1"}}},"commit":{"parents":[{"commit":"23287bc88e89984a0c11dd79731a1b4c02c9270a","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/23287bc88e89984a0c11dd79731a1b4c02c9270a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-10 23:56:15.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-10 23:56:15.000000000","tz":-420},"subject":"Fix deprecated glusterfs option","message":"Fix deprecated glusterfs option\n\n\"glusterfs_native_path_to_private_key\" is a deprecated\nreference to \"glusterfs_path_to_private_key\" and the\nfallback path is being removed [1]\n\n[1] https://review.opendev.org/745206/\n\nChange-Id: I47c795016c10f3afb7df4697ecf78880b8063b5a\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/02163665486062aabb53023c464e4706259850f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/02163665486062aabb53023c464e4706259850f9"}]},"branch":"refs/heads/master"},"2e453343e049cc960ff4a295ffa237b1a99b3687":{"kind":"REWORK","_number":2,"created":"2020-08-11 04:57:43.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/06/745606/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/06/745606/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/2"}}},"commit":{"parents":[{"commit":"23287bc88e89984a0c11dd79731a1b4c02c9270a","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/23287bc88e89984a0c11dd79731a1b4c02c9270a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-10 23:56:15.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-11 04:57:41.000000000","tz":0},"subject":"Fix deprecated glusterfs option","message":"Fix deprecated glusterfs option\n\n\"glusterfs_native_path_to_private_key\" is a deprecated\nreference to \"glusterfs_path_to_private_key\" and the\nfallback path is being removed [1]\n\n[1] https://review.opendev.org/745206/\n\nChange-Id: I47c795016c10f3afb7df4697ecf78880b8063b5a\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/2e453343e049cc960ff4a295ffa237b1a99b3687"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/2e453343e049cc960ff4a295ffa237b1a99b3687"}]},"branch":"refs/heads/master"},"9cb42f1c79e5b15ef585127cc9fd69dd86f46175":{"kind":"REWORK","_number":3,"created":"2020-08-11 19:46:33.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/06/745606/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/06/745606/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/3"}}},"commit":{"parents":[{"commit":"23287bc88e89984a0c11dd79731a1b4c02c9270a","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/23287bc88e89984a0c11dd79731a1b4c02c9270a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-10 23:56:15.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-11 19:46:17.000000000","tz":-420},"subject":"Fix deprecated glusterfs option","message":"Fix deprecated glusterfs option\n\n\"glusterfs_native_path_to_private_key\" is a deprecated\nreference to \"glusterfs_path_to_private_key\" and the\nfallback path is being removed [1]\n\n[1] https://review.opendev.org/745206/\nChange-Id: I47c795016c10f3afb7df4697ecf78880b8063b5a\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9cb42f1c79e5b15ef585127cc9fd69dd86f46175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9cb42f1c79e5b15ef585127cc9fd69dd86f46175"}]},"branch":"refs/heads/master"},"5ada9705d6a4ba53800327ca41fd622137c19134":{"kind":"REWORK","_number":4,"created":"2020-08-11 23:03:17.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/06/745606/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/06/745606/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/4"}}},"commit":{"parents":[{"commit":"23287bc88e89984a0c11dd79731a1b4c02c9270a","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/23287bc88e89984a0c11dd79731a1b4c02c9270a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-11 23:03:03.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-11 23:03:03.000000000","tz":-420},"subject":"Fix deprecated netapp option","message":"Fix deprecated netapp option\n\n\"netapp_root_volume_name\" is a deprecated\nreference to \"netapp_root_volume\" and the\nfallback path is being removed [1]\n\n[1] https://review.opendev.org/745206/\nChange-Id: I47c795016c10f3afb7df4697ecf78880b8063b5a\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/5ada9705d6a4ba53800327ca41fd622137c19134"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/5ada9705d6a4ba53800327ca41fd622137c19134"}]},"branch":"refs/heads/master"},"9f5b5de52c2e123c2046815d4400e43f00ecdd3b":{"kind":"REWORK","_number":5,"created":"2020-08-12 00:48:18.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/06/745606/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/06/745606/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/5"}}},"commit":{"parents":[{"commit":"23287bc88e89984a0c11dd79731a1b4c02c9270a","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/23287bc88e89984a0c11dd79731a1b4c02c9270a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-10 23:56:15.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-12 00:48:10.000000000","tz":-420},"subject":"Fix deprecated glusterfs option","message":"Fix deprecated glusterfs option\n\n\"glusterfs_native_path_to_private_key\" is a deprecated\nreference to \"glusterfs_path_to_private_key\" and the\nfallback path is being removed [1]\n\n[1] https://review.opendev.org/745206/\nChange-Id: I47c795016c10f3afb7df4697ecf78880b8063b5a\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9f5b5de52c2e123c2046815d4400e43f00ecdd3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9f5b5de52c2e123c2046815d4400e43f00ecdd3b"}]},"branch":"refs/heads/master"},"bacfbc1c661a089c879fb66b1a965231254a9e0d":{"kind":"REWORK","_number":6,"created":"2020-08-12 13:56:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/745606/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/06/745606/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/06/745606/6"}}},"commit":{"parents":[{"commit":"23287bc88e89984a0c11dd79731a1b4c02c9270a","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/23287bc88e89984a0c11dd79731a1b4c02c9270a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-10 23:56:15.000000000","tz":-420},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-12 13:54:57.000000000","tz":0},"subject":"Fix deprecated glusterfs option","message":"Fix deprecated glusterfs option\n\n\"glusterfs_native_path_to_private_key\" is a deprecated\nreference to \"glusterfs_path_to_private_key\" and the\nfallback path is being removed [1]\n\n[1] https://review.opendev.org/745206/\nChange-Id: I47c795016c10f3afb7df4697ecf78880b8063b5a\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/bacfbc1c661a089c879fb66b1a965231254a9e0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/bacfbc1c661a089c879fb66b1a965231254a9e0d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
