)]}'
{"id":"openstack%2Fpuppet-manila~746530","triplet_id":"openstack%2Fpuppet-manila~stable%2Fussuri~Id379b9aa91d6fc53dc2a36cd5d9504458c13734b","project":"openstack/puppet-manila","branch":"stable/ussuri","hashtags":[],"change_id":"Id379b9aa91d6fc53dc2a36cd5d9504458c13734b","subject":"Expose vhost_custom_fragment","status":"MERGED","created":"2020-08-17 14:54:12.000000000","updated":"2020-08-18 07:54:56.000000000","submitted":"2020-08-18 07:54:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746530-1597737296544-44e8c23f","meta_rev_id":"4930db035db22efdd1cf35fcbcc71f484704137f","_number":746530,"virtual_id_number":746530,"owner":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-18 07:54:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-17 21:52:48.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-18 07:00:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-18 07:00:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-17 21:52:48.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-18 07:00:50.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-18 07:54:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b8357b90976c47a5bf0ac40343c1f522e1a377bf","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-17 14:54:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d2f18ff3c42cbe60afb3f205c5ec17af1582ecc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 15:57:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f63f16bc8fdf40a0b934edf7088506ed : SUCCESS in 6m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a60f1b4190941f6ab376ef601d0d170 : SUCCESS in 6m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b980ccd143784244b1108cf1443e570c : SUCCESS in 5m 52s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/821a8e0949824187ae280abe87484227 : SUCCESS in 16m 04s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/cb83be86ba5d4aad805f7e82e3358bca : SUCCESS in 20m 00s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54809f1541cf4ea0aee052c4f331ea2a : SUCCESS in 11m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f3241f6ec3c44ed9f4145fbc8f96a87 : FAILURE in 8m 24s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4270db31f6bc461996cfda209b90d27b : SUCCESS in 56m 02s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2651f0ddca054bd5b4e56ef24dbc9782 : SUCCESS in 59m 16s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/dd998b31daf5467fba621c9c1fe7e13d : SUCCESS in 17m 45s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/0b488e1c1b4f44d1bdbf57cad8a16d2b : FAILURE in 11m 26s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3a94053460949adadc23b653822c5ca : FAILURE in 7m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b2f0f533106472a35569777f457dd7a18263e86","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-17 21:52:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ce6b772d564520d119baf2de4ac7af33a19c30b6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-18 07:00:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ef4f375f6b7743a9111aa6b83f7c2e0a35382b0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 07:01:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3732a910e48dd9fa54c52c43a22d913a7a54715d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 07:54:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4930db035db22efdd1cf35fcbcc71f484704137f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 07:54:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5221d7ffee24f9b8684e4994ea2cd16 : SUCCESS in 4m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c949f5503464411810984d0ab194001 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/741eb88ed38c49cb97ce35c4f0f62e6d : SUCCESS in 4m 34s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c26c9db3e6114174b865caee1a574480 : SUCCESS in 16m 21s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/001667d346f5477c8d92c79eabfe642a : SUCCESS in 13m 06s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6512521a87b84de281b16cac2ce08a8b : SUCCESS in 10m 42s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7b12387e962c471cb60191f2393d4183 : SUCCESS in 48m 49s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2f55620be2ad4166bc47d8ff0f32d53d : SUCCESS in 18m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"93646efae902cb709faeab2bd8fdc52f6b0f2dea","revisions":{"93646efae902cb709faeab2bd8fdc52f6b0f2dea":{"kind":"REWORK","_number":1,"created":"2020-08-17 14:54:12.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/30/746530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/30/746530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/30/746530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/30/746530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/30/746530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/30/746530/1"}}},"commit":{"parents":[{"commit":"e07189ffb6f6bf7a6086d34839712a5da9ed32ac","subject":"Switch to Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/e07189ffb6f6bf7a6086d34839712a5da9ed32ac"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-13 17:50:31.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-17 14:54:04.000000000","tz":-240},"subject":"Expose vhost_custom_fragment","message":"Expose vhost_custom_fragment\n\nIt is provided by the Puppet class \u0027openstacklib::wsgi::apache\u0027.\nThis change exposes it for the Manila service.\n\nChange-Id: Id379b9aa91d6fc53dc2a36cd5d9504458c13734b\nSigned-off-by: Luke Short \u003cekultails@gmail.com\u003e\n(cherry picked from commit 4fbe89816416ee80a86332f2ed2d22863c57816a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/93646efae902cb709faeab2bd8fdc52f6b0f2dea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/93646efae902cb709faeab2bd8fdc52f6b0f2dea"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
