)]}'
{"id":"openstack%2Fpuppet-manila~747525","triplet_id":"openstack%2Fpuppet-manila~master~Ifa605ced3f6cb1472535e0ce9dd776a4a47c0328","project":"openstack/puppet-manila","branch":"master","topic":"auth_type","hashtags":[],"change_id":"Ifa605ced3f6cb1472535e0ce9dd776a4a47c0328","subject":"Set credential parameters when auth_type !\u003d password","status":"MERGED","created":"2020-08-22 12:27:07.000000000","updated":"2020-08-24 16:44:52.000000000","submitted":"2020-08-24 16:42:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747525-1598287371429-69a35627","meta_rev_id":"27fd57b55367182a0b7dcc0b4dc82b02c5835c91","_number":747525,"virtual_id_number":747525,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-24 16:42:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-24 15:39:55.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-23 21:02:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-24 15:39:55.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-23 21:02:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-24 15:39:55.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-24 16:42:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bdc7de0694fbd7a7c7d50f36d362b9ab8af535a0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-22 12:27:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db30543e8c850218d77d4a2b0f04deb87cb7b9df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 13:16:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4c6d2093c624289ba2291f72f5a712a : SUCCESS in 4m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66553122951d48ae9dc4378cba1c6a66 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba26b7f93c3c4b12ba795d1a8a05def5 : SUCCESS in 4m 17s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/fa11740b902a4edda3e5546626a5947e : SUCCESS in 9m 19s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/00a2d664cc9c438c8fc7e067ffb20f61 : SUCCESS in 12m 55s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2c08b066dd94b87bf1b24389c3e2160 : SUCCESS in 12m 42s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11652f1d25dc4c3fa31cc403d54ae11a : FAILURE in 7m 36s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b36b281f2eef43cb9ea412e69fed04fc : SUCCESS in 41m 35s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1c5c6cdb233244a6962d99c77f9d2de1 : SUCCESS in 47m 43s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ad957b2277204526a94f6d7b1a5615f9 : SUCCESS in 19m 54s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/3cbd91b163ad4f7c9c6e5786372975bc : FAILURE in 10m 00s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6af1bbd629914b3b955abd0fa1ff1716 : FAILURE in 7m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f63dffcf1368439ab0509621af4bb1fd : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"16db35a4fccd239913d3778eab4ae12ad2127379","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-23 21:02:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6fe4eee1a77b56654f62a76c07291d7edc757e86","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-24 15:39:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5079abde93cedce805c4bb02e373347c66584db5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 15:40:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"56bae241faa01a9efadb58d4e528a34d4c05bf4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 16:42:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2dd6e55aa23323d95fecb179c0f002e62510d6db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 16:42:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1db50209a6e84cae9f2ea2e58d7da311 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/584c85d14b5e44cca131ed1c6c2eaec5 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b4a366feab74f259af8d52097c968c4 : SUCCESS in 4m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f17906a89fb54aceb7537015adfe48cb : SUCCESS in 13m 00s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1fd97b88f09c48d38ee3ced557e25521 : SUCCESS in 13m 42s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f81409c35897448083553e520ddb3c51 : SUCCESS in 11m 00s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e6d34bf284f94608acf9aacafb73ec46 : SUCCESS in 55m 13s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fe1a173b68804382be6076f1e4b5cdaf : SUCCESS in 17m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dffe7e50adf464f8771098b0ea16aca : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ef09bcb366bda20b2b329edee4a73adbdd9f80b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 16:44:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18c3c70119a7418d8e857afb738e7a8b : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"167e104ef519e457320a2c2cceed222338875493","revisions":{"167e104ef519e457320a2c2cceed222338875493":{"kind":"REWORK","_number":1,"created":"2020-08-22 12:27:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/747525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/25/747525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/25/747525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/25/747525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/25/747525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/25/747525/1"}}},"commit":{"parents":[{"commit":"fe08c435f01ee7ce9d11c5012d0bce96bead6df5","subject":"Merge \"Add support for db purge cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/fe08c435f01ee7ce9d11c5012d0bce96bead6df5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-22 12:26:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-22 12:26:25.000000000","tz":540},"subject":"Set credential parameters when auth_type !\u003d password","message":"Set credential parameters when auth_type !\u003d password\n\nCurrently some classes ignore keystone credential parameters like\nusername, but it causes the problem with some other \"auth_type\"s like\nv3password, which require these credential parameters.\n\nThis change makes sure that all credential parameters are set\nregardless of auth_type value.\n\nChange-Id: Ifa605ced3f6cb1472535e0ce9dd776a4a47c0328\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/167e104ef519e457320a2c2cceed222338875493"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/167e104ef519e457320a2c2cceed222338875493"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
