)]}'
{"id":"openstack%2Fpuppet-manila~756777","triplet_id":"openstack%2Fpuppet-manila~stable%2Fvictoria~I7d384047784cacf44d6c05a8d8e7b72a53a52bd9","project":"openstack/puppet-manila","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"I7d384047784cacf44d6c05a8d8e7b72a53a52bd9","subject":"Update .gitreview for stable/victoria","status":"MERGED","created":"2020-10-08 14:41:45.000000000","updated":"2020-10-12 18:15:21.000000000","submitted":"2020-10-12 18:15:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756777-1602526521365-ed338f59","meta_rev_id":"cf1f8a1803143f98ecd379b2c9f20b951d311c39","_number":756777,"virtual_id_number":756777,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-12 18:15:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-10-08 14:48:12.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-10-12 15:44:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-10-12 15:44:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 14:48:12.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-12 15:44:21.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-10-12 18:15:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89df18923f358542c7fe3ab9ca1c207b05923324","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-10-08 14:41:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e55954933dd65c2a50267f5c793df65a4cac2cb","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-08 14:48:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"47b293ee9f1b0a35992b845b02592620dd50a29c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 16:09:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ad8ba21eff04dbebf52269afca0a550 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a72f9497aa21477881dfe98a121e6677 : SUCCESS in 4m 09s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/414cefc66b1d46d3bd278f098e3d0627 : SUCCESS in 12m 05s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f5e99666c4f42a7a7e4d438111288bd : SUCCESS in 9m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dad82da9897c449080aa8ba195002e86 : SUCCESS in 10m 39s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4ef7d914a61e4efa9f3cbf619801fdc6 : SUCCESS in 43m 52s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5d5c6fd2326b4f5e83cb9c984bafa72b : SUCCESS in 58m 34s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f5539aac22ab4aa6bb019e1fbc332e7f : FAILURE in 9m 45s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/229a1fe25bbf433db51a373364f2d668 : FAILURE in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"412c0c027374f83e8c2f8247ee15d633abb496da","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-12 15:44:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c5be527195cf0b2b38dcd7851021412b2366eb60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 17:11:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc1f7a17e0574bbeadebf9c2f0390d45 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28e1a0879c444ef3a17e9658defa49de : SUCCESS in 3m 24s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/781c8f85ad264260a73c69b530035c5c : SUCCESS in 14m 43s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce76bedba0cb446094bad4df5b5ab4ed : SUCCESS in 9m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f386b652713d472baa2fd72fe689599f : SUCCESS in 10m 55s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4b66cce337e74efabe56fa961545dcb9 : SUCCESS in 46m 33s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d3393dbce8944cf081e4cc6bd3bb2d9a : SUCCESS in 59m 27s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/e21abea6988849c38450380e098f707d : SUCCESS in 21m 08s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2144d031e73d4a579a0594c4d391dc10 : SUCCESS in 15m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3fa3e05ddf696054ce4649d31155ee95c70ecf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 17:12:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa2272213904d80f142392baa84063275001ee3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 18:15:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8a8bec8b0fbd8edf14496aa9c24403e0738611e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 18:15:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/794bfed50e2e44f2afd512047d3ae6f9 : SUCCESS in 3m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5b84138a9c8406c84271a8f275869c8 : SUCCESS in 4m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9f1256f89fa342b3b325b04f3fa60add : SUCCESS in 13m 07s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff93c7d5119c4ef88a3a843862fa7906 : SUCCESS in 10m 20s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/be599fa47a7c4346b3545705a2e00c1c : SUCCESS in 59m 06s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/8a22f8bb504741f8a4bf0fd10b7448dc : SUCCESS in 19m 52s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88f8b927ce734e91b1c2b3dc6a4dbce0 : SUCCESS in 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9efc9a8549dfa9430acccaddc8635879f3017e6","revisions":{"f9efc9a8549dfa9430acccaddc8635879f3017e6":{"kind":"REWORK","_number":1,"created":"2020-10-08 14:41:45.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/77/756777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/77/756777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/77/756777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/77/756777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/77/756777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/77/756777/1"}}},"commit":{"parents":[{"commit":"9db569ed766f6d58891e3e50df0fd0cf34ae01e6","subject":"Prepare Victoria RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9db569ed766f6d58891e3e50df0fd0cf34ae01e6"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-10-08 14:41:44.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-10-08 14:41:44.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: I7d384047784cacf44d6c05a8d8e7b72a53a52bd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f9efc9a8549dfa9430acccaddc8635879f3017e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f9efc9a8549dfa9430acccaddc8635879f3017e6"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
