)]}'
{"id":"openstack%2Fpuppet-manila~756967","triplet_id":"openstack%2Fpuppet-manila~master~Ib56008c22658d93a1aaaf494ae4ef0524ff36a97","project":"openstack/puppet-manila","branch":"master","topic":"missing-deps","hashtags":[],"change_id":"Ib56008c22658d93a1aaaf494ae4ef0524ff36a97","subject":"Include deps class in unit test for mysql","status":"MERGED","created":"2020-10-09 02:00:42.000000000","updated":"2020-10-09 03:13:16.000000000","submitted":"2020-10-09 03:13:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756967-1602213196454-a063aa70","meta_rev_id":"ac1920e5ceeee90078ac56d987d15506456020ae","_number":756967,"virtual_id_number":756967,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-09 03:13:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-10-09 02:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-10-09 02:47:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 03:13:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"034bab3fe52815c94e24f0e14873b1d2642f4830","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-09 02:00:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b14192dc394b7d12200f1f68357a7d6debcb821","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 02:21:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1954c2ec2d54af69e8d5d2625f61396 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4437d5bc227b4aa287727aa6ebaf21bf : SUCCESS in 3m 48s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4dca2c123a4a49cca07562116e09c5ec : FAILURE in 11m 47s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9b20233d4954f01b06fef755b4ef81a : FAILURE in 9m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2379302e4a548698cf193248d84071f : FAILURE in 10m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0ddeff1befcb39aa89cb1f4b09924550338f3ce","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-09 02:27:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"848a1e83ad4272c4c8d565ece3c83ffe086ddd51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 02:46:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7030b4f56c4e4ed89a6479d045e1d0ea : SUCCESS in 3m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/501498f2a4274fdf82422343f2c95b65 : SUCCESS in 4m 53s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c089784f94914474bfddb8f68f18ee2e : SUCCESS in 11m 29s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b73fcb6f03b479c879a2d57b32b5fff : SUCCESS in 9m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/815a71a12285429d88923ae914ea9d2d : SUCCESS in 10m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ed45177999af693bc989aa78fdde4e2bcfc3000a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-09 02:47:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"70b921adccc8247ece387df57b8910546fbeb535","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 02:47:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"11a652cbc92b79049d648d2886a570a01f9b947f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 03:13:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2a1a0d908ee8ee243b04db9764b803235d34c531","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 03:13:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4958a747e0d48b4abf47b874c9c2daa : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09e7529640b74323a83e16af795ce49a : SUCCESS in 3m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8677af3ccebd44c0bc471870978c2818 : SUCCESS in 16m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/433bc8ecab4a4d74a31249a9df58f54d : SUCCESS in 9m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d0ef1c47d6eefdb204f7cf5faa2de4c572201cb1","revisions":{"7dd950f28c8be43b69f41adbbee4dea01c0e7467":{"kind":"REWORK","_number":1,"created":"2020-10-09 02:00:42.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/67/756967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/67/756967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/67/756967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/67/756967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/67/756967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/67/756967/1"}}},"commit":{"parents":[{"commit":"9db569ed766f6d58891e3e50df0fd0cf34ae01e6","subject":"Prepare Victoria RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9db569ed766f6d58891e3e50df0fd0cf34ae01e6"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-09 01:59:55.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-09 01:59:55.000000000","tz":480},"subject":"Include deps class in unit test for mysql","message":"Include deps class in unit test for mysql\n\nChange-Id: Ib56008c22658d93a1aaaf494ae4ef0524ff36a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/7dd950f28c8be43b69f41adbbee4dea01c0e7467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/7dd950f28c8be43b69f41adbbee4dea01c0e7467"}]},"branch":"refs/heads/master"},"d0ef1c47d6eefdb204f7cf5faa2de4c572201cb1":{"kind":"REWORK","_number":2,"created":"2020-10-09 02:27:37.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/67/756967/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/67/756967/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/67/756967/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/67/756967/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/67/756967/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/67/756967/2"}}},"commit":{"parents":[{"commit":"9db569ed766f6d58891e3e50df0fd0cf34ae01e6","subject":"Prepare Victoria RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9db569ed766f6d58891e3e50df0fd0cf34ae01e6"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-09 01:59:55.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-09 02:27:23.000000000","tz":480},"subject":"Include deps class in unit test for mysql","message":"Include deps class in unit test for mysql\n\nChange-Id: Ib56008c22658d93a1aaaf494ae4ef0524ff36a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/d0ef1c47d6eefdb204f7cf5faa2de4c572201cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/d0ef1c47d6eefdb204f7cf5faa2de4c572201cb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
