)]}'
{"id":"openstack%2Fpuppet-manila~758907","triplet_id":"openstack%2Fpuppet-manila~master~I5227b94592868f8cd8f482779fea22225a1ecfeb","project":"openstack/puppet-manila","branch":"master","topic":"init-db","hashtags":[],"change_id":"I5227b94592868f8cd8f482779fea22225a1ecfeb","subject":"Deprecate database options from the manila class","status":"MERGED","created":"2020-10-21 00:34:23.000000000","updated":"2020-10-21 09:17:43.000000000","submitted":"2020-10-21 09:16:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758907-1603271763584-ea0a8f3d","meta_rev_id":"22495e4b01e7428eba642bd28e9dde0232ffa96c","_number":758907,"virtual_id_number":758907,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-21 09:16:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-10-21 05:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-10-21 07:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2020-10-21 07:41:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-21 05:26:54.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-10-21 07:41:57.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-10-21 09:16:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30b9114c59b4621c642e0282dc319f2c8d8829f4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-21 00:34:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66108fc60954e5a22f2aeb86f268989a8dc5dead","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-21 00:35:02.000000000","message":"Topic set to init-db","accounts_in_message":[],"_revision_number":1},{"id":"4c4a1113c7e03466c3414dc67baf2536769b1404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 01:39:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b56f85a0fb2c4cfab1f253327c1ec876 : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e8ae6b717e845edb13f93ad6ba3b9ba : SUCCESS in 4m 31s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/749cd1163d6e450b927cc6fc722165b7 : SUCCESS in 13m 51s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/187ae5c95ac64a28860ba6befd26c3c3 : SUCCESS in 9m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7921664d9c664b38abe2a4811b36fd3f : SUCCESS in 11m 17s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8ec05912786440bd9a5bdf59a8c82c34 : SUCCESS in 42m 38s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d82ea79ac7fd47d4a5d66e793de5188b : SUCCESS in 1h 00m 30s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/11db9a61346243f3b16b10a75423a944 : SUCCESS in 23m 35s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a5cc62c7694450289a8a57a8a5d5629 : SUCCESS in 17m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f8ddb08f07d4c33ae07aab6224a436d : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d37f5d543c6bce3fe8ed4bbf77b880be9e5d2fa0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-21 05:26:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"85faf62594358abf26f19fb52e58993066848f3b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-21 07:41:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0fb30c958878d454fd748931dff77ec562338419","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 07:42:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4fbee657701843a629ae68af5a7397386a54f719","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 09:16:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/509fbb60735b4784aa9341f356f5a0ff : SUCCESS in 6m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/450d396812a24898a4d61c133c4b185e : SUCCESS in 6m 30s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f8281c0eb67048a1b4d335088481d69d : SUCCESS in 15m 07s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0078bf4f0a344af7a61c61458557973f : SUCCESS in 12m 37s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a24ed1a22b964ba98bd79cd773b9a5dd : SUCCESS in 1h 02m 47s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c8eb571b15834881b481a84e544c0668 : SUCCESS in 22m 37s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db29adf0bed24542846a507e0a9b05eb : SUCCESS in 18m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f78861574387445c84bcd5df9dfb1edb : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":1},{"id":"57d8000c8ec9daa3b3412d0718c4b3d0dd9147cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 09:16:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"22495e4b01e7428eba642bd28e9dde0232ffa96c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 09:17:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cbbc5e2c59846c7a42b7d9e758ec370 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12a789cddc8b062cef0544d4b80821cbba8cda0f","revisions":{"12a789cddc8b062cef0544d4b80821cbba8cda0f":{"kind":"REWORK","_number":1,"created":"2020-10-21 00:34:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/758907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/07/758907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/07/758907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/07/758907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/07/758907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/07/758907/1"}}},"commit":{"parents":[{"commit":"9db569ed766f6d58891e3e50df0fd0cf34ae01e6","subject":"Prepare Victoria RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9db569ed766f6d58891e3e50df0fd0cf34ae01e6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-20 03:06:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-20 03:06:04.000000000","tz":540},"subject":"Deprecate database options from the manila class","message":"Deprecate database options from the manila class\n\n... because these options have been implemented in the manila::db\nclass.\n\nChange-Id: I5227b94592868f8cd8f482779fea22225a1ecfeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/12a789cddc8b062cef0544d4b80821cbba8cda0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/12a789cddc8b062cef0544d4b80821cbba8cda0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
