)]}'
{"id":"openstack%2Fpuppet-manila~765269","triplet_id":"openstack%2Fpuppet-manila~stable%2Fussuri~I71ef2b5831420a59a3ef0fb9dbad672c27734c1d","project":"openstack/puppet-manila","branch":"stable/ussuri","topic":"at-interface-stable/ussuri","hashtags":[],"change_id":"I71ef2b5831420a59a3ef0fb9dbad672c27734c1d","subject":"Add support for the interface parameter in authtoken middleware","status":"MERGED","created":"2020-12-03 06:35:38.000000000","updated":"2020-12-04 07:35:52.000000000","submitted":"2020-12-04 07:34:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765269-at-interface-stable/ussuri","cherry_pick_of_change":740947,"cherry_pick_of_patch_set":1,"meta_rev_id":"0281ab4500e4244e71ac2cac97009b50e54d17ae","_number":765269,"virtual_id_number":765269,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-04 07:34:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"recommended":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"all":[{"value":1,"date":"2020-12-03 12:09:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},{"value":2,"date":"2020-12-04 06:39:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-12-03 21:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},{"value":1,"date":"2020-12-04 06:39:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-03 06:35:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"state":"REVIEWER"},{"updated":"2020-12-03 08:49:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 21:59:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-12-04 06:39:39.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"5822ca47c0c3e408967b3438ed1019ce4cc2b596","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-03 06:35:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"458889c45789257c525bd3b3062f81255fdeee6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-03 06:44:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"97e79ccb67b888cafa933064e2d889583abc576a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 08:49:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b66a1d54dfcb4c2392d90e223f7d7581 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/084cc9ccbda049259c7894018b14e634 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8c03b3d1ee7450ab3d34d1904f53b0d : SUCCESS in 3m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/23e76cabe25142c1a5d414ae5cf4de50 : SUCCESS in 14m 21s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/65846016596e4a5ea4f110eb2effef25 : SUCCESS in 14m 25s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1403dcaabde04b7cbf8a9b8d1aba972b : SUCCESS in 11m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/896f9835cb48412b975dd755561aaf69 : SUCCESS in 10m 23s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a459c0985f2449fa803bf28cb8a5caaa : SUCCESS in 37m 34s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/54a51fb8872b448e99819b4379658379 : SUCCESS in 58m 25s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/3e430e7b738841f8b21e924f3fe8b010 : SUCCESS in 19m 31s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34fd8c870762468faee6d6ce70c8fbec : SUCCESS in 16m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbb2754e274b439b8c3e484ba03c5ecb : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"5edae3fad7d47679d021b151c6fb29a48c0e76fe","author":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"date":"2020-12-03 12:09:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"90d3ccf85c28b2abf1efc6456913734e6e32c2c1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-12-03 21:59:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2f6f44307e904b03a336f580d4ad96a89ab3ad90","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-12-04 06:39:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c03985db8d764c427f895b242ffd03b0493f1105","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 06:40:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a918a29ae542b7d6821759ff81ab5b81606fd4a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 07:34:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec31e94deba4447b9650711d00617ab1 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d86f098579d242ccb874f3640d323577 : SUCCESS in 7m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b04affcad2d431ca43c12280fc1e82a : SUCCESS in 2m 31s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/1c43ddd4092c463394918438a4f4fd18 : SUCCESS in 13m 48s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/8bbeefc8b0144b6a866fb1b94af72d47 : SUCCESS in 10m 22s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efd57d9641814263b4b49e2a9a05392d : SUCCESS in 10m 15s\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fb0d83ddd5894191af849804365f3b5f : SUCCESS in 44m 17s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/4a0c8df563fd432592e57c975f3620db : SUCCESS in 18m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f6542c002034af185f9760eed9e847d : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"be9ba8bc3b1fbcb4a175bbaa0b31a86d2dc3e853","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 07:34:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0281ab4500e4244e71ac2cac97009b50e54d17ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 07:35:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cd42ea557d342b080699bc3539434bf : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"16253272242c403c73e69c41ece49fb4fbb98f3c","revisions":{"7a69b9f358a20b6cbdd013b7002c76a08c2df593":{"kind":"REWORK","_number":1,"created":"2020-12-03 06:35:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/765269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/69/765269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/69/765269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/69/765269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/69/765269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/69/765269/1"}}},"commit":{"parents":[{"commit":"a307316a538aceba6c2f6db5cf2391933de8b412","subject":"Use common class to manage apache service in beaker job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/a307316a538aceba6c2f6db5cf2391933de8b412"}]}],"author":{"name":"Lewis Denny","email":"ldenny@redhat.com","date":"2020-07-14 12:16:10.000000000","tz":600},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-03 06:35:38.000000000","tz":0},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: I71ef2b5831420a59a3ef0fb9dbad672c27734c1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/7a69b9f358a20b6cbdd013b7002c76a08c2df593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/7a69b9f358a20b6cbdd013b7002c76a08c2df593"}]},"branch":"refs/heads/stable/ussuri"},"16253272242c403c73e69c41ece49fb4fbb98f3c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-03 06:44:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/765269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/69/765269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/69/765269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/69/765269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/69/765269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/69/765269/2"}}},"commit":{"parents":[{"commit":"a307316a538aceba6c2f6db5cf2391933de8b412","subject":"Use common class to manage apache service in beaker job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/a307316a538aceba6c2f6db5cf2391933de8b412"}]}],"author":{"name":"Lewis Denny","email":"ldenny@redhat.com","date":"2020-07-14 12:16:10.000000000","tz":600},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-03 06:44:50.000000000","tz":0},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: I71ef2b5831420a59a3ef0fb9dbad672c27734c1d\n(cherry picked from commit b92f8a9238b981f18f88540adcd3957efd4bdc81)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/16253272242c403c73e69c41ece49fb4fbb98f3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/16253272242c403c73e69c41ece49fb4fbb98f3c"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
