)]}'
{"id":"openstack%2Fpuppet-manila~770083","triplet_id":"openstack%2Fpuppet-manila~stable%2Fussuri~I5b6c85b7d7a865a0f281a725f8df8b8f0e3d12e8","project":"openstack/puppet-manila","branch":"stable/ussuri","topic":"ignored-ut-stable/victoria-stable/ussuri","hashtags":[],"change_id":"I5b6c85b7d7a865a0f281a725f8df8b8f0e3d12e8","subject":"Fix ignored unit tests for manila::volume::cinder","status":"ABANDONED","created":"2021-01-11 11:54:50.000000000","updated":"2021-01-11 12:31:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":769368,"cherry_pick_of_patch_set":1,"meta_rev_id":"b22c01e3c274febcebcc405fca4faba08418377e","_number":770083,"virtual_id_number":770083,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-11 12:24:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-11 12:24:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"115d5ba260dac3639873cdf3b19689f5f16114a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-11 11:54:50.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"56db7d5ddcf860a076a4a701318897478bd4a5e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 12:24:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afc9d5dfd8874e29a9b358c9e8d3d9eb : SUCCESS in 3m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb1081e6db6b4ab3ae723a49b96ab972 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca80836092b64c32b52e75da4a20aae8 : SUCCESS in 3m 58s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/58bba494617c4df6bc158a9b2c6fe7d9 : FAILURE in 11m 35s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/2334b7ebf01f4618b6ecd8a49510ee7f : FAILURE in 8m 29s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5e5e56c082b4675ba9b42dd7c74ef17 : FAILURE in 10m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82a8065d6cfe48e598c1f137e1818d72 : FAILURE in 8m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b22c01e3c274febcebcc405fca4faba08418377e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-11 12:31:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5756649c67bdeb5e6c25965a8aeafd1cda18f69","revisions":{"f5756649c67bdeb5e6c25965a8aeafd1cda18f69":{"kind":"REWORK","_number":1,"created":"2021-01-11 11:54:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/770083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/83/770083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/83/770083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/83/770083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/83/770083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/83/770083/1"}}},"commit":{"parents":[{"commit":"16253272242c403c73e69c41ece49fb4fbb98f3c","subject":"Add support for the interface parameter in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/16253272242c403c73e69c41ece49fb4fbb98f3c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-03 05:29:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-11 11:54:50.000000000","tz":0},"subject":"Fix ignored unit tests for manila::volume::cinder","message":"Fix ignored unit tests for manila::volume::cinder\n\nUnit test files should be named like *_spec.rb.\n\nChange-Id: I5b6c85b7d7a865a0f281a725f8df8b8f0e3d12e8\n(cherry picked from commit fea6aabfe2ef829d255e296f966141597eface1e)\n(cherry picked from commit ba0babc70be998de7141d0cd9de84fc84767af26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f5756649c67bdeb5e6c25965a8aeafd1cda18f69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f5756649c67bdeb5e6c25965a8aeafd1cda18f69"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
