)]}'
{"id":"openstack%2Fpuppet-manila~805950","triplet_id":"openstack%2Fpuppet-manila~master~I5cb24d4a556fcba5b4d4e4bf87570b95adb6b76e","project":"openstack/puppet-manila","branch":"master","hashtags":[],"change_id":"I5cb24d4a556fcba5b4d4e4bf87570b95adb6b76e","subject":"Fix the lvm_share_export_ips typo","status":"MERGED","created":"2021-08-25 06:47:17.000000000","updated":"2021-09-03 14:53:28.000000000","submitted":"2021-09-03 14:50:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805950","meta_rev_id":"cf765ad8e1afcfed771ee1ffd423c3508f8195a7","_number":805950,"virtual_id_number":805950,"owner":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 14:50:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-09-02 23:34:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-09-03 13:42:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-09-03 13:42:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-25 07:57:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-25 16:59:04.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-09-02 16:00:59.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-02 23:11:04.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"bd71f42f245961032c58db48e908a2da23761042","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-08-25 06:47:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a1b8deec1306d74723386c4e8fa7ba1980e1ad0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-08-25 06:49:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a831ddb68f2fa0fc21093ed93beb02a3737bfbd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 07:57:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbccda1840eb4e618f57bf4843cd93b4 : SUCCESS in 5m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2e8c7decdcc4225b3149e470bc0e187 : SUCCESS in 5m 51s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d3328d55cf84bcb9bd939a0d820bca1 : SUCCESS in 5m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c12a5a7b51f43229302d3eeaa07c40b : SUCCESS in 14m 33s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/507f0684d138446f954699e6c7e6cd7e : SUCCESS in 15m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d7bdb93b27f47409ff0b2c2a90f61ff : SUCCESS in 12m 37s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be95e162ab684943b1bf4dd5f1fcebbd : SUCCESS in 12m 40s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d926763d07e14dd1838e0cb4be5c7f9c : SUCCESS in 12m 22s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/695e663bfac54c2482d1ba0f9319e617 : SUCCESS in 44m 49s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d069cdb846d4fa2ba6c5db715aef05f : SUCCESS in 1h 02m 33s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f5fd90309f94df2a1a452de5ee62c22 : FAILURE in 19m 32s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9cd8b7d055b4754b534168a55ff87e8 : SUCCESS in 23m 38s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e016f140ba474955aac34553f062e77e : FAILURE in 18m 56s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3edb3e699d494295bb16a5ead079a969 : FAILURE in 18m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"17b46ff2b0509228c52c716fa6900e84283fbd0d","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2021-08-25 16:59:04.000000000","message":"Patch Set 2: Code-Review+1\n\nYes, [1] confirms it should be lvm_share_export_ips.\n\nIs there a Launchpad bug for this? Having one (and referencing it in this patch) will facilitate backporting the fix.\n\n[1] https://opendev.org/openstack/manila/src/branch/master/manila/share/drivers/lvm.py#L45","accounts_in_message":[],"_revision_number":2},{"id":"c46887c39eba45bfc66ce7cfc265b94c2bdad354","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-08-26 00:38:49.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review+1\n\u003e \n\u003e Yes, [1] confirms it should be lvm_share_export_ips.\n\u003e \n\u003e Is there a Launchpad bug for this? Having one (and referencing it in this patch) will facilitate backporting the fix.\n\u003e \n\u003e [1] https://opendev.org/openstack/manila/src/branch/master/manila/share/drivers/lvm.py#L45\n\nThanks for confirming.\nIt is not a bug, just misuse deprecated parameter in the previous change. lvm_share_export_ip is deprecated in favor of lvm_share_export_ips https://opendev.org/openstack/manila/src/branch/master/releasenotes/notes/drop-support-for-lvm-share-export-ip-e031ef4c5f95b534.yaml","accounts_in_message":[],"_revision_number":2},{"id":"bdbf38cc8adc50944f1d14a8972ba154b41213b0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-02 16:00:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d71a2c25d8e9071843a5c57864daeda953b36116","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-02 23:11:04.000000000","message":"Patch Set 2: Code-Review-1\n\nCan we deprecate the parameter instead of directly rename it ?","accounts_in_message":[],"_revision_number":2},{"id":"4443843858e4a6aaf62e31674e97bc63c4a8b4d6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-02 23:28:49.000000000","message":"Patch Set 2:\n\nHmm... so this resource type was added during this cycle. So it might be just redundant to deprecate the parameter.\nA small concern is that we created a release with the wrong parameter already, and adding a release note would be helpful. I\u0027ll add that one by my side soon.","accounts_in_message":[],"_revision_number":2},{"id":"7895f6ebdb9e03e3e2b42652188cf49234851215","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-02 23:33:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a5c568edcf61e93f10a14d62214f3812a5b4d5a9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-02 23:34:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fe2d2fd426cc60bbd8bcb03938a253a5836ebef6","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-09-02 23:48:21.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Hmm... so this resource type was added during this cycle. So it might be just redundant to deprecate the parameter.\n\u003e A small concern is that we created a release with the wrong parameter already, and adding a release note would be helpful. I\u0027ll add that one by my side soon.\n\nThanks Takashi","accounts_in_message":[],"_revision_number":3},{"id":"985719cd018a1f84fd4a4aa8271c9155d4af0df6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 00:48:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a8bc70dceb34ce0bd2a5429130dce1a : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57af3932593c411390bbe3ea22655889 : SUCCESS in 4m 12s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e4e6ff2ea6e4222b5dd0a139bf1b792 : SUCCESS in 3m 59s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/802309b20ef040d58a4fec2f50116bfc : SUCCESS in 12m 37s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3e2c5cfeb4646b9a77323678fda70d1 : SUCCESS in 12m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/926c0b5fc28e40b19ccdf3b4ddee56ca : SUCCESS in 8m 30s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aee5d59384ae4b1a9363356e49e73a10 : SUCCESS in 9m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93e1d35a8e3841cab1e697590e1c22aa : SUCCESS in 9m 04s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9301f33ac02492fb9ffd26c43c55ac3 : SUCCESS in 37m 43s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/080feb60c64b4032a61261c1594657ce : SUCCESS in 1h 03m 37s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17b68f4005a64e62959ea6f523aa8df9 : FAILURE in 16m 56s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6691ece9649b455bb073f7ca78748a68 : SUCCESS in 21m 13s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5bcccf512414459890daef2a73aaf15 : FAILURE in 35m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b43d8bc20e2d40819e2e801bcc2517b7 : FAILURE in 34m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b5c4f30218c4571af70923e6468dafe : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":3},{"id":"23de1312d438cebefc81c9107dba8accf4aa82f4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-03 13:42:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"73bfdc2a937836991b364bd66364ab7615e39690","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-03 13:42:42.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"144efdb40b9b5de13792aead450785b681be2337","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 13:43:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"347ae29c2de6e0e38669cb27a6bebbbdd4f95c0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 14:50:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33653b4c4c70450da4548d541537488d : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a20c049622544ff8a0ef494554783560 : SUCCESS in 4m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c86122c0e7354371a873b703bad07c49 : SUCCESS in 4m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/245b89a9ab1b4e608bd2190e1a3e4d0e : SUCCESS in 13m 09s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2096500ffac04aa99068ae5473f2a7a6 : SUCCESS in 15m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d412cdbbf1b44e34a01711370b05c581 : SUCCESS in 10m 54s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13ea6a68b1994f29aa4028ea30c4836d : SUCCESS in 13m 13s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/96563723cdf94cbea50db65bda759d6a : SUCCESS in 1h 00m 15s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/29bcbb83ed4d4b88898b8607590c0e62 : SUCCESS in 28m 13s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7227e15526824c9d970da5321bc44775 : FAILURE in 18m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/306176fb2f084e1a86146c71ed5bf58e : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":3},{"id":"4b1087c4a1d6739b87e155abf07e94cf59d4cad0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 14:50:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"cf765ad8e1afcfed771ee1ffd423c3508f8195a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 14:53:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac2b9c891ade4a7fa53cf4c6760baef1 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fcfffb1fcae33981953369e0bb9050abbeb67e01","revisions":{"0cf10dbe2628b0c40099e1e2aa1b18769b111ea9":{"kind":"REWORK","_number":1,"created":"2021-08-25 06:47:17.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/50/805950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/50/805950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/805950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/805950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/805950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/50/805950/1"}}},"commit":{"parents":[{"commit":"152321b77cb10d090fb4cac3de4d454ae664446a","subject":"Merge \"Remove deprecated support for cephfs_enable_snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/152321b77cb10d090fb4cac3de4d454ae664446a"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2021-08-25 06:46:13.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2021-08-25 06:46:13.000000000","tz":600},"subject":"Fix the lvm_share_export_ips typo","message":"Fix the lvm_share_export_ips typo\n\nChange-Id: I5cb24d4a556fcba5b4d4e4bf87570b95adb6b76e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/0cf10dbe2628b0c40099e1e2aa1b18769b111ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/0cf10dbe2628b0c40099e1e2aa1b18769b111ea9"}]},"branch":"refs/heads/master"},"f1c8b44c0c3b089a618ab7be662674d9917b2e2c":{"kind":"REWORK","_number":2,"created":"2021-08-25 06:49:59.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/50/805950/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/50/805950/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/805950/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/805950/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/805950/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/50/805950/2"}}},"commit":{"parents":[{"commit":"152321b77cb10d090fb4cac3de4d454ae664446a","subject":"Merge \"Remove deprecated support for cephfs_enable_snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/152321b77cb10d090fb4cac3de4d454ae664446a"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2021-08-25 06:46:13.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2021-08-25 06:49:34.000000000","tz":600},"subject":"Fix the lvm_share_export_ips typo","message":"Fix the lvm_share_export_ips typo\n\nChange-Id: I5cb24d4a556fcba5b4d4e4bf87570b95adb6b76e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f1c8b44c0c3b089a618ab7be662674d9917b2e2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f1c8b44c0c3b089a618ab7be662674d9917b2e2c"}]},"branch":"refs/heads/master"},"fcfffb1fcae33981953369e0bb9050abbeb67e01":{"kind":"REWORK","_number":3,"created":"2021-09-02 23:33:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/805950/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/50/805950/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/805950/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/805950/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/805950/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/50/805950/3"}}},"commit":{"parents":[{"commit":"152321b77cb10d090fb4cac3de4d454ae664446a","subject":"Merge \"Remove deprecated support for cephfs_enable_snapshots\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/152321b77cb10d090fb4cac3de4d454ae664446a"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2021-08-25 06:46:13.000000000","tz":600},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-02 23:32:53.000000000","tz":540},"subject":"Fix the lvm_share_export_ips typo","message":"Fix the lvm_share_export_ips typo\n\nChange-Id: I5cb24d4a556fcba5b4d4e4bf87570b95adb6b76e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/fcfffb1fcae33981953369e0bb9050abbeb67e01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/fcfffb1fcae33981953369e0bb9050abbeb67e01"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
