)]}'
{"id":"openstack%2Fpuppet-manila~855300","triplet_id":"openstack%2Fpuppet-manila~master~Iffebfe17f535a24975a55a158a7123c1ce73aebf","project":"openstack/puppet-manila","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-01 09:39:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iffebfe17f535a24975a55a158a7123c1ce73aebf","subject":"DNM: Test missing privse-helpers in manila-sudoers","status":"ABANDONED","created":"2022-08-31 00:22:39.000000000","updated":"2022-09-01 09:39:41.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"922c663e317c73ac63ef7cf9df98dc0ff275336c","_number":855300,"virtual_id_number":855300,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-01 04:13:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-31 01:00:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1be6ae90e6f13f3cd923bab758f4d836899e7a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-31 00:22:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06287fab0da1031dc0acbb29ef1cc27c01918568","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 01:00:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2adc7458d02e4ebb92e28c8cbeb3a2a9\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b59c12b28b5a4f29bdcd953a5335ead7 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c97054d8f404497bc8b6bf3986a0d43 : SUCCESS in 3m 14s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c64e82669364c8baee4e96393f1670c : SUCCESS in 3m 07s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a596637853f04a72b882dad0ffec99a5 : FAILURE in 10m 23s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/77b42512a8904b4d91eaf4f4265c9014 : FAILURE in 30m 56s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f054d7d97379435bb088c32877136e95 : FAILURE in 14m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a64a7145abb485d8f5b114684c100b8 : FAILURE in 15m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fda09d682f25545204bd41a8379b140b0017b2eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-31 01:22:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d5eed32cd5b86ebf782aef84fbba3ee6c3a07075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 01:36:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/86acfd3a371842568712e072e8279822\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24315cd023db4204b767f43eb3e2b09a : SUCCESS in 3m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49e8bc092689432d9d44b7c7b3006ddd : SUCCESS in 2m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1dcf841791f4481e8e3d5e348b96133f : SUCCESS in 2m 59s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21f3457781b648078cc42a4cc1837747 : FAILURE in 6m 28s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/73e675eebc7d4f6a812e142f26eb6d17 : FAILURE in 12m 40s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc013d0732214870939698472608dd12 : FAILURE in 6m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b36f5a65a894bb5b48ad7df1ea1dd2e : FAILURE in 6m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab1ef3d482ffee5220fc9f38a3671ef6c3320ff7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-31 01:46:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"67e1552b34cd032ccb9b80dc04b2198583648c7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 02:03:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eef6c94a8d0d43d492c10b313c97913b\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9c41cf89c5c4b6db08e89050d591a37 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9059515ee29402fae284bca825355d4 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbdc855ec21b48d98aee947437260b50 : SUCCESS in 2m 43s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4627120287c49b08627373c449ce29c : FAILURE in 6m 27s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b0b6f95dd1f454ebcc1dbb6e4dd92d8 : FAILURE in 11m 08s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8583dc9f0c04fdaa78e059b30308df8 : FAILURE in 6m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f8b9d4f819c45f5bdb91f83cf8bbd0e : FAILURE in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"73980c69b94eb860918b549204d1dd39a2b68017","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-31 03:26:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"17095f3cfbf1af1b669a86ff66efae79de42b01d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-31 03:42:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4f87cd326accd9202de520e06fd7b3247a5cd201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 04:11:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a754f0ed45d3430cb04733a7f54aaec3\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a3d1e4303a94c5e98e7278316da868f : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49affb8a66ff4027a8436d51103181e0 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec1e3936bae340069fa2b79fe36ec836 : SUCCESS in 4m 04s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6fb7301565a4d808b773a33f19af0e8 : FAILURE in 6m 45s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/6fea679a09ae4d84b0635eb23e34dccb : SUCCESS in 25m 41s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f938cba97e69420d95eec8404d6faa8f : FAILURE in 19m 02s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d16595c43b5043d3a388ccefd69222ef : FAILURE in 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c09e292860b7154acdd0a675d129567d9ba7073b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-31 04:42:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bdd6a2ebd9857f6e1e9e9fe09b93d1dbd4cbf9fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 05:04:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ead0ccfdba649be8c6d0988d1b835eb\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd284722bf3c46e48f93e53927d13c88 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7568b31394346fba952258c0f1314ae : SUCCESS in 3m 29s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/387eb418e9394789afe2ccc667b2d3d7 : SUCCESS in 2m 52s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9213e2a26eca4855a79a4db096690b52 : FAILURE in 9m 56s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff63eec124e64a4e876fa9d15375df11 : SUCCESS in 20m 51s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/159a74007f484d98a4df24484d276864 : FAILURE in 15m 57s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/021d76956c744b588c0e0dd54e05c478 : FAILURE in 15m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e0621c9d3b6c9cda31a5eb65bc59aa26c9755a9f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-31 13:15:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"227151bbb6807f9c20708ebff98923c6242768b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 13:36:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a10aee1405be488eb0ad7e3c4df7ca3e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69736fb4589f4c61b8c769ccde7d023f : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71ff1caa1d9b4408bf9bf71d823c4d04 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29fa801a8beb42f5b5f27261d5b04cf1 : SUCCESS in 3m 15s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4be5cb849f9b4ca2884eecc971d8f21f : FAILURE in 9m 04s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ade5e3f551a74bfc91078aaa37e6a7a5 : SUCCESS in 20m 00s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1adae70f292749a28615418c8c12d195 : FAILURE in 14m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c701d117731241d4abbb639670396fc2 : FAILURE in 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f910c3a9877b806417efae7a7d2e9bf09a29c501","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-31 15:39:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b1089a8e31dbae6deec40d143e6f9905b0f4224f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 16:09:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0e173b840554386b50ea7d439072742\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ae967dd31a04254b9e4bbad9e9e0dda : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83c71eaa28314b8a8c66206cec3f042d : SUCCESS in 4m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6129e69b9b54575a2a422ee1a13f3f6 : SUCCESS in 4m 14s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aae71a8640a44e31beb2894fe112c95b : FAILURE in 12m 17s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa46abe493914cc2bd3059e322dbbe55 : SUCCESS in 25m 37s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c135c803d8ab46d8ae9c2c2483bc2815 : FAILURE in 19m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf40d023974c4c3dbf070a7d6130784e : FAILURE in 26m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"64b3c7f1deffd4a57699ef588868bd5970034bdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-01 00:12:35.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"efdb5678c1cd2f5badbaccac77da13059221f2b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 00:27:26.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb203254ca444001bbb23558380e56e9\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d90fe65353504c37bbf54929b2075e7f : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e6641a2571244179cce9780a0318123 : SUCCESS in 3m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/853b247ccd43452b90e27ae64196a375 : SUCCESS in 3m 05s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eec4c38a4a59465eb0a0f11ec587703f : FAILURE in 9m 24s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/8565eb869ab344dca33308bf5420d90a : FAILURE in 13m 15s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc85b8475f5548a89081912bdabec856 : FAILURE in 9m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36fc0c4f9ee542c7a9e760fef80a0004 : FAILURE in 7m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f7e8e5cb9717361829ca4ac134baaf9321981719","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-01 00:48:38.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"4828f251b9417a6bbd6eaf857f0f6c82f29756fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-01 00:48:54.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"d8b07770005f847ec12b77955e75d98e0f0facd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 01:16:32.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e153d522f8ed4867abca2503a32ba940\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5e800ceea404c19b4c15e328e03001a : SUCCESS in 3m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c574e7c484884d56b6cafbf3ab2dda70 : SUCCESS in 3m 08s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5c3356d3d8840faa6aa96b5688a1ec5 : SUCCESS in 3m 49s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53fd7cae963f4ce9ac89185ffb1f3026 : FAILURE in 8m 48s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7e36bf648a44fb784e97cce69ae40dc : SUCCESS in 25m 56s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6b5115d67284acaa1d12ecc48805c75 : FAILURE in 16m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/486e64c38eda449e9b19e4dcfdfe32f5 : FAILURE in 18m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"38b938171647d8aaa34edc75bfee8105be43ad01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-01 01:39:24.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"dca384d602772eddf458fa1ebd62dfae28a6b0ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-01 01:39:35.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"efe79a57cd0e93f41717101438ea2512e11bc721","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-01 01:48:59.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"d0d2eb48b2b03536cfb8aca48d77b9b67cdad73f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 02:17:20.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d667ca5bc694fdcb94d78bf5cb98aba\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43a57640b038428d9eacc51b0dcd5069 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6ce152059cf4c1aa8e2cff4922e2162 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00cd6e05b57e4082beec8a41adebcfa5 : SUCCESS in 3m 15s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54c73b418d914498a767fc0b71ee1e81 : FAILURE in 17m 32s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9c5a921458b4f2faa9fc589597a669c : SUCCESS in 25m 24s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df784a3200d843c5b67fca002bfd8302 : FAILURE in 19m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca7b1ba3be8b4277874c9ef744450807 : FAILURE in 14m 51s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b1c110ee1ff7b9bd51c25b38a1dd6682e2f73fbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-01 02:36:50.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"71bf3d4f0cdae3022dba5f2ce145927fd6e05f15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 03:01:41.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1522506ed1eb438ea10dccb7d63f0ecb\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad14183637134c08bf39ae1037d6852b : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d492957b45e8429fa3bc08cd3216f819 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45028a6c81af4e29ba977b4b29cce33b : SUCCESS in 4m 03s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a34b3fa0cfd412d9e26f093c80f76ba : FAILURE in 7m 39s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/68f76608d87b4733a6a083820c570d4d : SUCCESS in 21m 14s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a10491af803c443b9d0467a1141ca8aa : FAILURE in 17m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edcbc948f72142fd9856ea927e42f7f8 : FAILURE in 20m 24s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a9c54a61f0553ddc8a92e51affb9eb1de2954097","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-01 03:45:21.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"c2280b3c808ec9c6e160c3960fb4a659d7a631e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 04:13:04.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d28604ad6c14102909cd6d5f2da043e\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7783953f68554a40b04241091063a4c8 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3609ccc51bfc4ffd95b44eb91bfefb38 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/627cf2fc6c3b45bd9fbb81e899c3123e : SUCCESS in 3m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d867b7f156b4803aaa020645b7c10a6 : FAILURE in 9m 44s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/6daf0f68ab2040d1baaa73a724333617 : SUCCESS in 21m 56s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3874a68164ca4153af6cfcd048192a61 : FAILURE in 16m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/779d0d8cc9c944e6af7b71b87ec9bc86 : FAILURE in 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"922c663e317c73ac63ef7cf9df98dc0ff275336c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-01 09:39:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"a769284f6a70413ce1a4f424c834e537411746a8","revisions":{"9861a3198247084796ae8ed0e6e419a00c6bcc37":{"kind":"REWORK","_number":1,"created":"2022-08-31 00:22:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/1"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"subject":"DNM: Test missing rootwrap filter rule for Manila","message":"DNM: Test missing rootwrap filter rule for Manila\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9861a3198247084796ae8ed0e6e419a00c6bcc37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9861a3198247084796ae8ed0e6e419a00c6bcc37"}]},"branch":"refs/heads/master"},"d4c6581b9657fa8a5015cdbca8496bf5831e0ed3":{"kind":"REWORK","_number":2,"created":"2022-08-31 01:22:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/2"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 01:22:11.000000000","tz":540},"subject":"DNM: Test missing rootwrap filter rule for Manila","message":"DNM: Test missing rootwrap filter rule for Manila\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/d4c6581b9657fa8a5015cdbca8496bf5831e0ed3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/d4c6581b9657fa8a5015cdbca8496bf5831e0ed3"}]},"branch":"refs/heads/master"},"0e2c87b76eba2b68bb4434717cb5d6fcd8ad9bd7":{"kind":"REWORK","_number":3,"created":"2022-08-31 01:46:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/3"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 01:46:40.000000000","tz":540},"subject":"DNM: Test missing rootwrap filter rule for Manila","message":"DNM: Test missing rootwrap filter rule for Manila\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/0e2c87b76eba2b68bb4434717cb5d6fcd8ad9bd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/0e2c87b76eba2b68bb4434717cb5d6fcd8ad9bd7"}]},"branch":"refs/heads/master"},"a5e0271af5133502edbdf7bb9f99554e6da9a10b":{"kind":"REWORK","_number":4,"created":"2022-08-31 03:26:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/4"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 03:26:40.000000000","tz":540},"subject":"DNM: Test missing rootwrap filter rule for Manila","message":"DNM: Test missing rootwrap filter rule for Manila\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/a5e0271af5133502edbdf7bb9f99554e6da9a10b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/a5e0271af5133502edbdf7bb9f99554e6da9a10b"}]},"branch":"refs/heads/master"},"a42c255cebe6c854addee59376aa4c3a8c9ac4e2":{"kind":"REWORK","_number":5,"created":"2022-08-31 03:42:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/5"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 03:42:30.000000000","tz":540},"subject":"DNM: Test missing rootwrap filter rule for Manila","message":"DNM: Test missing rootwrap filter rule for Manila\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/a42c255cebe6c854addee59376aa4c3a8c9ac4e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/a42c255cebe6c854addee59376aa4c3a8c9ac4e2"}]},"branch":"refs/heads/master"},"1705fd65269ae95c2348e1840e2a6ac7b5279b2d":{"kind":"REWORK","_number":6,"created":"2022-08-31 04:42:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/6"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 04:41:57.000000000","tz":540},"subject":"DNM: Test missing rootwrap filter rule for Manila","message":"DNM: Test missing rootwrap filter rule for Manila\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1705fd65269ae95c2348e1840e2a6ac7b5279b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1705fd65269ae95c2348e1840e2a6ac7b5279b2d"}]},"branch":"refs/heads/master"},"f75f1d7c05a84b7e129d1c329f0c3e7430bc0d43":{"kind":"REWORK","_number":7,"created":"2022-08-31 15:39:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/7"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 15:39:41.000000000","tz":540},"subject":"DNM: Test missing rootwrap filter rule for Manila","message":"DNM: Test missing rootwrap filter rule for Manila\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f75f1d7c05a84b7e129d1c329f0c3e7430bc0d43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f75f1d7c05a84b7e129d1c329f0c3e7430bc0d43"}]},"branch":"refs/heads/master"},"58a3fe7305d361aa34d16bc7e5e88a76a4cf445c":{"kind":"REWORK","_number":8,"created":"2022-09-01 00:12:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/8"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-01 00:12:04.000000000","tz":540},"subject":"DNM: Test missing privse-helpers in manila-sudoers","message":"DNM: Test missing privse-helpers in manila-sudoers\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/58a3fe7305d361aa34d16bc7e5e88a76a4cf445c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/58a3fe7305d361aa34d16bc7e5e88a76a4cf445c"}]},"branch":"refs/heads/master"},"014dcb48f0a93031bb8442a358df5e2b12de0c81":{"kind":"REWORK","_number":9,"created":"2022-09-01 00:48:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/9"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-01 00:48:37.000000000","tz":0},"subject":"DNM: Test missing privse-helpers in manila-sudoers","message":"DNM: Test missing privse-helpers in manila-sudoers\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/014dcb48f0a93031bb8442a358df5e2b12de0c81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/014dcb48f0a93031bb8442a358df5e2b12de0c81"}]},"branch":"refs/heads/master"},"49fde1eca3520fb793c7f94258562e1081e90649":{"kind":"REWORK","_number":10,"created":"2022-09-01 00:48:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/10"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-01 00:48:54.000000000","tz":0},"subject":"DNM: Test missing privse-helpers in manila-sudoers","message":"DNM: Test missing privse-helpers in manila-sudoers\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/49fde1eca3520fb793c7f94258562e1081e90649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/49fde1eca3520fb793c7f94258562e1081e90649"}]},"branch":"refs/heads/master"},"02c96d1a6a55bed3a906f97bca80d685777439f0":{"kind":"REWORK","_number":11,"created":"2022-09-01 01:39:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/11"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-01 01:39:24.000000000","tz":0},"subject":"DNM: Test missing privse-helpers in manila-sudoers","message":"DNM: Test missing privse-helpers in manila-sudoers\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/02c96d1a6a55bed3a906f97bca80d685777439f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/02c96d1a6a55bed3a906f97bca80d685777439f0"}]},"branch":"refs/heads/master"},"3a98c511475a4237d976eff389d9ac832430d5ee":{"kind":"REWORK","_number":12,"created":"2022-09-01 01:39:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/12"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-01 01:39:35.000000000","tz":0},"subject":"DNM: Test missing privse-helpers in manila-sudoers","message":"DNM: Test missing privse-helpers in manila-sudoers\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/3a98c511475a4237d976eff389d9ac832430d5ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/3a98c511475a4237d976eff389d9ac832430d5ee"}]},"branch":"refs/heads/master"},"46ac0d59c07acbc2e95367e1697701ffe296961e":{"kind":"REWORK","_number":13,"created":"2022-09-01 01:48:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/13"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-01 01:48:59.000000000","tz":0},"subject":"DNM: Test missing privse-helpers in manila-sudoers","message":"DNM: Test missing privse-helpers in manila-sudoers\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/46ac0d59c07acbc2e95367e1697701ffe296961e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/46ac0d59c07acbc2e95367e1697701ffe296961e"}]},"branch":"refs/heads/master"},"19b4ba2583690fa9af089f4090a189b7f7e6ffef":{"kind":"REWORK","_number":14,"created":"2022-09-01 02:36:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/14"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-01 02:36:50.000000000","tz":0},"subject":"DNM: Test missing privse-helpers in manila-sudoers","message":"DNM: Test missing privse-helpers in manila-sudoers\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/19b4ba2583690fa9af089f4090a189b7f7e6ffef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/19b4ba2583690fa9af089f4090a189b7f7e6ffef"}]},"branch":"refs/heads/master"},"a769284f6a70413ce1a4f424c834e537411746a8":{"kind":"REWORK","_number":15,"created":"2022-09-01 03:45:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/855300/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/00/855300/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/00/855300/15"}}},"commit":{"parents":[{"commit":"819ca23648d8db1727c2ed7756dc0ce9a50d1cb7","subject":"WIP: Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/819ca23648d8db1727c2ed7756dc0ce9a50d1cb7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 00:22:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-01 03:45:20.000000000","tz":0},"subject":"DNM: Test missing privse-helpers in manila-sudoers","message":"DNM: Test missing privse-helpers in manila-sudoers\n\nChange-Id: Iffebfe17f535a24975a55a158a7123c1ce73aebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/a769284f6a70413ce1a4f424c834e537411746a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/a769284f6a70413ce1a4f424c834e537411746a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
