)]}'
{"id":"openstack%2Fpuppet-manila~855720","triplet_id":"openstack%2Fpuppet-manila~master~I481ba2a60a2aa2b7b4a2470f07b5f858db85d065","project":"openstack/puppet-manila","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-08 05:45:14.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I481ba2a60a2aa2b7b4a2470f07b5f858db85d065","subject":"Use p-o-i manifest in acceptance tests","status":"MERGED","created":"2022-09-03 19:29:08.000000000","updated":"2022-09-08 06:45:11.000000000","submitted":"2022-09-08 06:45:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855720","meta_rev_id":"517b538ca1acd1d8862a22d8b10aa3fa4c246db6","_number":855720,"virtual_id_number":855720,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-08 06:45:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2022-09-06 16:40:21.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2022-09-06 18:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-09-07 17:44:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-09-07 17:44:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-03 19:56:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-06 16:40:21.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2022-09-06 18:08:19.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-09-07 17:44:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"c8d4b30e315f74b8a1e3282333da394c4a3f147d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-03 19:29:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99e045d360b8b523dbb78b0440b7c1a7d54cc1dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 19:56:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b770b964347d4f93af349978aae43818\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be399e4568c24faaaa52075266289d1f : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8a892ff1f4e4984a557fbc9cdd4a3eb : SUCCESS in 3m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af23dacd68f54548b2e8c5e41e57e22b : SUCCESS in 4m 26s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a2f3ec92f984b92857f791dfa14a29d : SUCCESS in 16m 28s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/63021aeacc1d4b75b45c1c87954ab38c : SUCCESS in 25m 25s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fbef74db9cc4e3886dfd73c66279950 : SUCCESS in 16m 56s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df531e40a2be468e8b3a0e664538037a : SUCCESS in 13m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f251bd7b3a5a41e3d561e4a6c70b8f9258b17b96","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-09-06 16:40:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e990d9adf4b92c2fd04714e02c5142c195528bdb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-06 18:08:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb9f01e73487c2e1fc4244a9f99582ca26b952dd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-09-07 17:44:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b06a3da4603e960c193ccd35457dad0b894cf077","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 17:44:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/935be84c9d684f50a26189f36fee5eaf\n\n\nWarning:\n  Change 855535 in project openstack/puppet-openstack-integration does not share a change queue with 855720 in project openstack/puppet-manila","accounts_in_message":[],"_revision_number":1},{"id":"d09925d7e6b4015f03c8a7b88b8718775f535eb7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-08 05:45:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f6ca8444a7ec0964faf55a83c13fb1214afd9bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 06:14:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55ed5f947d644015ad4db9bfdfaa0694\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d348bff2afdc49b684d26e0d9e4f88a9 : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7db92d6fc4554511b36eb2c49f0ca90e : SUCCESS in 4m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91f534ecf19147a3ad1db391204c5e44 : SUCCESS in 3m 58s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d1b21527ef6465ab2ec5db7a59b7227 : SUCCESS in 17m 38s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/590ba52dbe65404bb123766f13b4388b : SUCCESS in 23m 59s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2926f01518d14a11b43c5243806e9eaa : SUCCESS in 17m 18s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a4a8ad698f04187bd47c005bef82277 : SUCCESS in 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71dc63f1f4cce8d3c8f7efd05fec03ec4b18c14f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 06:15:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f94830672a377bceb8684013c9b34c574b9b0eb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 06:45:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9898428ae554755a326e98545408184\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c48b34684bb143c0b1a976c6eaf61ec4 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e87fa9577e7e4531bb17d5ef18cc8937 : SUCCESS in 3m 13s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14a10e8f3b15478b8286d2d2f77dae50 : SUCCESS in 3m 14s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e81e21871ca4f34b561b426ae16005b : SUCCESS in 24m 17s","accounts_in_message":[],"_revision_number":1},{"id":"517b538ca1acd1d8862a22d8b10aa3fa4c246db6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 06:45:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d18a8e3741d212d9a0c9f51f6257f7562ebb6ee1","revisions":{"d18a8e3741d212d9a0c9f51f6257f7562ebb6ee1":{"kind":"REWORK","_number":1,"created":"2022-09-03 19:29:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/855720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/20/855720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/20/855720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/20/855720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/20/855720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/20/855720/1"}}},"commit":{"parents":[{"commit":"715e97a3bf4b50a7ecbf92ec91c216a637cf7e4d","subject":"Enable manila-share in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/715e97a3bf4b50a7ecbf92ec91c216a637cf7e4d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-03 19:28:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-03 19:28:43.000000000","tz":540},"subject":"Use p-o-i manifest in acceptance tests","message":"Use p-o-i manifest in acceptance tests\n\nDepends-on: https://review.opendev.org/855535\nChange-Id: I481ba2a60a2aa2b7b4a2470f07b5f858db85d065\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/d18a8e3741d212d9a0c9f51f6257f7562ebb6ee1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/d18a8e3741d212d9a0c9f51f6257f7562ebb6ee1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
