)]}'
{"id":"openstack%2Fpuppet-manila~860513","triplet_id":"openstack%2Fpuppet-manila~master~I52f81faf2008e6d8c152503ca2d706fd962b8ed3","project":"openstack/puppet-manila","branch":"master","topic":"revert-system-scope","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-10-12 00:11:18.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I52f81faf2008e6d8c152503ca2d706fd962b8ed3","subject":"Do not use system scope tokens in providers","status":"MERGED","created":"2022-10-06 01:36:31.000000000","updated":"2022-10-12 02:39:04.000000000","submitted":"2022-10-12 02:37:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860513-revert-system-scope","meta_rev_id":"bfe71438f6fd37c7635a46f81a7308161059fb7f","_number":860513,"virtual_id_number":860513,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-12 02:37:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-10-07 04:32:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-10-11 09:28:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-10-11 12:37:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-10-11 09:28:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-06 02:38:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-07 04:32:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-10-11 09:28:43.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-10-11 12:37:10.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"614a2ef3ea4485915b42c790727afbab2e352098","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-06 01:36:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1143c95d8be291bc356adf1a253e410478701f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 02:38:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9ee4f7753944e6ba6adfe895a05f6f0\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62a1c787b2cd4c4c8fbd10ee599b32ea : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b07e35feef354edcb0c56bd10d79c318 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2995728c6daf43ebaf0591f88c0f6bd8 : SUCCESS in 3m 43s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/857aacd83b4d4417a7d95ea1bd4e0323 : SUCCESS in 13m 25s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f61eb0cf8e6440e902ee1fe6dc40806 : SUCCESS in 10m 41s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3522645ffff48b38e746f489064166a : SUCCESS in 10m 42s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bba2c1b3a9684da9b950241ef0b93849 : SUCCESS in 10m 32s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f766d98c8db422aa1b1d845e6a5c412 : SUCCESS in 33m 11s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/22a18fd6b5114728b7620cff7f45973a : SUCCESS in 59m 38s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc480e07fb71475c8ef51e82ffd5681b : SUCCESS in 13m 31s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e5c5b22dbe64dc0b102c6b14b640412 : SUCCESS in 20m 32s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dd9a8712ed34c0c9ac0be769f84e486 : SUCCESS in 17m 22s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b04d4cead604b67a73c54b3f78852f7 : SUCCESS in 13m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a9869d319f94b83b4e2f91e901ea2a7 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1d0106f0ea27b0a0fd6934d479344b1b67bf26d7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-06 03:00:03.000000000","message":"Topic set to revert-system-scope","accounts_in_message":[],"_revision_number":1},{"id":"d41ad8233cfdbb0726789bc98d1fba87a40f47d6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-10-07 04:32:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"032c60a9cda12778b2de4c9736cc57cbe40d8e18","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-11 09:28:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac3a31f90c532e48b30654a70ed4c5e16b7900fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 09:29:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a8243e78e53f524775141bb7b957358d684d586","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 10:40:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/147b76c0b4b94edf999095a46611c253\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bfbdb3b758343e7b252d6ea48acaec4 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ab871c57e4a48caac6a3562e8a31ff9 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9817d67dc4654470abff760def125c76 : SUCCESS in 3m 32s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a9832fa5c554229bfce022a022220e0 : SUCCESS in 9m 53s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1802975bc1f9402599fef105bbe11268 : SUCCESS in 9m 57s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c97809707daf4ceca937a75a2c7257c8 : SUCCESS in 8m 06s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d32ccdab85f4d07bfe4e0f865d85c1f : FAILURE in 1h 04m 11s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/59d1c96077424ccbb1bb33025efe0be1 : SUCCESS in 20m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1551c7d4204141a888b08576377b4fbf : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e796a3d187bbb39bb4e5a4de5ad5604dd284f6aa","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-10-11 12:37:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"971c03ad073f35e0daffc2e745ed4539f0000791","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-12 00:11:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e5dbba6022cb19b03b15fd9382683e8cf349597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 01:41:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a9ed5d513e44e9fb1f974ae43a745ae\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52bcab7f9eda46e89ae7d660750a6f6a : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78e932879a0f45adbb7bcb4c57bfcd89 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f69f84cb2b2448a99e439341640cde1 : SUCCESS in 3m 47s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/35b703ba31b544eaa86e989249550831 : SUCCESS in 9m 32s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6253b63d7e04b3385c4ebab35036075 : SUCCESS in 50m 39s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ec334d25d854cc8b34d80258d5f9051 : SUCCESS in 6m 28s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9d39954f7b342c6985d824df6303090 : SUCCESS in 7m 09s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bc001a275e8445890131734de03af74 : SUCCESS in 1h 27m 43s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/093e98da621344b5bbca22760557cea6 : SUCCESS in 39m 03s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d470bc81cde84077914eac52a211a56c : SUCCESS in 12m 20s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/42c4e010162349f7ad279c1a4d25a962 : SUCCESS in 22m 50s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2c87b1a43204f548c98c60517696169 : SUCCESS in 14m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72c9be329cf4472ea7d72ec617d203ff : SUCCESS in 15m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50eda00c6305457badee140a47a33911 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3c6ba5153b6b0b755d65fa1496a25ee046e29e32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 01:42:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35527077fa4fb5943d2c696bc268f4bc2bf04d32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 02:37:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/debe899d9cea49d99b1d721a2056b605\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/360da4962d6f4a948823b27ec393068d : SUCCESS in 3m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29339e494d18490f881bb3d4d3da1e65 : SUCCESS in 2m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bd8ebbd3e8046cbae2d3bb464ddbe26 : SUCCESS in 3m 08s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c205896f7764da3bf5ab865c805bccd : SUCCESS in 13m 03s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b6faf01c87e4cb9991b622378633135 : SUCCESS in 9m 40s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd8ba3f0f1d94db1bf2dbbd8ca61e604 : SUCCESS in 9m 10s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/76f2b97388be4aceb3bcb2ad1569b33a : SUCCESS in 54m 11s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/008072c2989246f298608f2cf7ec5bc6 : SUCCESS in 20m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26b3b04fc607427490002cde2fe6e481 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d4e18a5e9aef5910713c6e53aa7ec788a50c2b73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 02:37:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bfe71438f6fd37c7635a46f81a7308161059fb7f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 02:39:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bef4164b628f4dfcaa32cd3ae78f5e45\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2778dcec321047f29a0a4edd1460b18e : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6dd3edfe3df5d3e99ffc49333dd0784a292b7ca","revisions":{"a6dd3edfe3df5d3e99ffc49333dd0784a292b7ca":{"kind":"REWORK","_number":1,"created":"2022-10-06 01:36:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/13/860513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/13/860513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/13/860513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/13/860513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/13/860513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/13/860513/1"}}},"commit":{"parents":[{"commit":"d65fa22282eb5164b983ddae3a1236dce61f2120","subject":"Enable scenario 005 integration job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/d65fa22282eb5164b983ddae3a1236dce61f2120"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-06 01:31:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-06 01:36:22.000000000","tz":540},"subject":"Do not use system scope tokens in providers","message":"Do not use system scope tokens in providers\n\nThis is partial revert of 5ca6e6fc9cb2403307114877ece24e394be35794 .\n\nAfter discussing several problems caused by scope separation, we\ndecided to suspend implementing the scope enforcement and focus on\nproject personas like reader role. As the result of that decision,\nthe system admin persona will be removed, thus we should use\nthe project admin persona instead. The previous policy rules to allow\nsystem scope access have been reverted by [1].\n\n[1] 755a1503187a29f9b4f6ecbf369acb781c3e95e7\n\nChange-Id: I52f81faf2008e6d8c152503ca2d706fd962b8ed3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/a6dd3edfe3df5d3e99ffc49333dd0784a292b7ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/a6dd3edfe3df5d3e99ffc49333dd0784a292b7ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
