)]}'
{"id":"openstack%2Fpuppet-manila~862018","triplet_id":"openstack%2Fpuppet-manila~stable%2Fzed~I957351358eb70512a5816fddb5402044c9b3232d","project":"openstack/puppet-manila","branch":"stable/zed","topic":"create-zed","hashtags":[],"change_id":"I957351358eb70512a5816fddb5402044c9b3232d","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","status":"MERGED","created":"2022-10-20 11:06:53.000000000","updated":"2022-10-20 14:30:02.000000000","submitted":"2022-10-20 14:29:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862018-create-zed","meta_rev_id":"49628d0e2c80075b9babdd02a32e3932cd99449e","_number":862018,"virtual_id_number":862018,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-20 14:29:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-10-20 14:20:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-10-20 14:20:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-20 11:19:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 14:20:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"cf653ea505a36aad40e320e952ece73141f17eec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-10-20 11:06:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84c409138a984f81ff1735117ba1c254ad4aed7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 11:19:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b133fd155ad44cd6b24fdd3cadff5c32\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aae04626e0b54f2f8783c3bdd34cd29c : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e712b17410d4540b2255cfcbcac627a : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6148263b3946b86366e672e1b8a06dcc6087e134","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-20 14:20:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"598c73775e591c30b7146642cb6274484864c507","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 14:21:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee9fdc41c8a6b8ceb61ac709076278416ea1e3b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 14:29:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b2d1e1dfac14bb1a944397b1ef3ada8\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ed5de3e679949529592a1fa0db04de5 : SUCCESS in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4110a21abb74e59bb3b386a306cd6ed : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d77d10d098723e09f389b8a3c584fb7de938f658","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 14:29:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"49628d0e2c80075b9babdd02a32e3932cd99449e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 14:30:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20f9788a731348fb8f2a8bd8c72cb3f0\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/daeabdca88dc4653a7c57decd18b3a44 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12f66e81d184bc056f111f7bd0e466231400ced0","revisions":{"12f66e81d184bc056f111f7bd0e466231400ced0":{"kind":"REWORK","_number":1,"created":"2022-10-20 11:06:53.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/18/862018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/18/862018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/18/862018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/18/862018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/18/862018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/18/862018/1"}}},"commit":{"parents":[{"commit":"3115c1e2fa3adc9a977a73429028959ecd43b1d1","subject":"Update .gitreview for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/3115c1e2fa3adc9a977a73429028959ecd43b1d1"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-10-20 11:06:52.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-10-20 11:06:52.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","message":"Update TOX_CONSTRAINTS_FILE for stable/zed\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/zed branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I957351358eb70512a5816fddb5402044c9b3232d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/12f66e81d184bc056f111f7bd0e466231400ced0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/12f66e81d184bc056f111f7bd0e466231400ced0"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
