)]}'
{"id":"openstack%2Fpuppet-manila~869897","triplet_id":"openstack%2Fpuppet-manila~master~I405ee83fd03d2bdfee819d5f7a452c8477a9f9ed","project":"openstack/puppet-manila","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-01-12 16:08:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I405ee83fd03d2bdfee819d5f7a452c8477a9f9ed","subject":"hnas: Remove deprecated unused parameter","status":"MERGED","created":"2023-01-12 06:25:32.000000000","updated":"2023-01-12 16:10:31.000000000","submitted":"2023-01-12 16:08:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869897","meta_rev_id":"f8d7240279a49108f75ce0222c8058926535d4e5","_number":869897,"virtual_id_number":869897,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 16:08:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2023-01-12 14:05:05.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2023-01-12 14:52:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2023-01-12 14:52:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-12 07:32:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-12 14:05:05.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2023-01-12 14:52:07.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"4322fecd7255c40ce40efd9e5bfeb02f9fec4424","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-12 06:25:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"599bdd3a4a02654c208415376e925a801b77e6b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 07:32:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a080048d382479093ce733e8d81843c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c68b6ea3af7466ca6208df6946b4f8b : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63a9f8c91fbc468bbb47456bd4036779 : SUCCESS in 3m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c6d7ebc7c034620b280cb00610c2b5d : SUCCESS in 3m 34s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/a629f264c9824104b7233efcd44df2a2 : SUCCESS in 12m 59s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caceff34cc7c4583a5ee3f95bf739dfc : SUCCESS in 7m 48s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb08e89f011341cd9059a7f06e76c728 : SUCCESS in 7m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd7768149645414b8d72bd4da1499b97 : SUCCESS in 8m 01s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3067f95adfb4ed5a94e410d758db6a2 : FAILURE in 31m 03s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8800145ebb9f477f8ba6c16e2b78ee6f : SUCCESS in 1h 01m 07s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a37fb1c6f2248a287b548e23836eb8c : SUCCESS in 14m 15s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/12dda63fe9744ad4a44d66812b2616ee : SUCCESS in 18m 34s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c8abc3e89364ca4b933c393ea6dd4de : SUCCESS in 13m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14238ff439844b2fb9a3d646aad6de18 : SUCCESS in 13m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b514c9b2ceba49f884260fbc9b6c721b : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2e084ff367f2f53e7211a92f8459e308ee1543a4","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-01-12 14:05:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"85ac7d75cfc8e4410245fe33421148f47ea19449","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-01-12 14:52:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae90c2ca3784c1b9a179c7beb99d41bbc2950215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:53:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01e5ac0d0c8e6a9f0f37b8767e105889becf8ed8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 16:08:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4df2302cb68e4f989e9aa8e50b639f30\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9061f715408e47488737c418b30097ba : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d5dc7e8f6564fec922449e3d87cef53 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/092fa425b07245939c36b16c54e9796c : SUCCESS in 2m 57s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8f3d61dd2424ffc820087a37c5e8657 : SUCCESS in 11m 09s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ce550a018834f5d8b13ef87ad536875 : SUCCESS in 10m 05s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb9008db76e0440c81d9ade252c4bf22 : SUCCESS in 7m 54s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf2e990d4c414ac6a100791ed04b9603 : SUCCESS in 55m 28s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8c1d59859484d61b5b3d7e38a73a96b : SUCCESS in 21m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42ceaa3cc0ec4d0297a0a51dd2e9d3ce : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"02668f5312001b5a7e8033ba5d5de37b9b5a14d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 16:08:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f8d7240279a49108f75ce0222c8058926535d4e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 16:10:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d19f15630b044308caa256c53e63661\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a89abd1e77fc4a17b61139e6d53fe17a : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bad1441aad14b6d0f726ad9461fe397bfcc20db7","revisions":{"bad1441aad14b6d0f726ad9461fe397bfcc20db7":{"kind":"REWORK","_number":1,"created":"2023-01-12 06:25:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/869897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/97/869897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/97/869897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/97/869897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/97/869897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/97/869897/1"}}},"commit":{"parents":[{"commit":"959da1064ada645036df2c14a6fad65608bf5ba0","subject":"NetApp: Remove deprecated netapp_root_volume_name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/959da1064ada645036df2c14a6fad65608bf5ba0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-12 06:23:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-12 06:25:22.000000000","tz":540},"subject":"hnas: Remove deprecated unused parameter","message":"hnas: Remove deprecated unused parameter\n\nThe manila::share::hitachi_hnas::share_backend_name parameter has been\nunused and was formally deprecated during the Zed cycle[1] so can be\nremoved now.\n\nAlso, the wrong example is fixed by this change.\n\n[1] 49291198a3d230e9117f2420c99a38670efe25f8\n\nChange-Id: I405ee83fd03d2bdfee819d5f7a452c8477a9f9ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/bad1441aad14b6d0f726ad9461fe397bfcc20db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/bad1441aad14b6d0f726ad9461fe397bfcc20db7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
