)]}'
{"id":"openstack%2Fpuppet-manila~873479","triplet_id":"openstack%2Fpuppet-manila~master~I8d71ac0ab9fa4a97bc1f5df9a274209198183ddb","project":"openstack/puppet-manila","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-13 18:09:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8d71ac0ab9fa4a97bc1f5df9a274209198183ddb","subject":"Remove deprecated manila::service_instance","status":"MERGED","created":"2023-02-13 03:18:21.000000000","updated":"2023-02-13 18:10:39.000000000","submitted":"2023-02-13 18:09:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873479","meta_rev_id":"edb389b20d7c58a68abf911f63eca06225cb4ff9","_number":873479,"virtual_id_number":873479,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-13 18:09:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-02-13 15:38:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-02-13 17:12:04.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-02-13 17:12:04.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 04:17:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 15:38:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-02-13 17:12:04.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"1472ff50d30c60654823063745d4fdac7a65c9b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-13 03:18:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"754931ebe33a84e4524a74c2db324e84d3d08181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 04:17:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6508f931d37b480a919cd6b5b0f9bf7c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/949af002a78643c3b56c377ac0cb9802 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a7075f142174ad2bead545e91738a39 : SUCCESS in 2m 20s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ec079e4441c4c609379a323594ecd66 : FAILURE in 9m 41s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16388cb7bcc9485481933ac523d310ec : FAILURE in 7m 31s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87f12dde97e24e229646820ac590a9df : FAILURE in 9m 22s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64942042278b434484bc36ace77214a0 : FAILURE in 41m 06s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/692920411e064112b4202b04c41d2e0b : SUCCESS in 52m 43s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb00bea55cbf45dd8c8654418173ee58 : SUCCESS in 22m 20s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1983e8dbabc24e48a3c4fbe7b0b29369 : SUCCESS in 15m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8544f3384f2c45bda87dabf7df464484 : SUCCESS in 14m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/208ad0cf52724da19aed54b00a3a9fcb : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"bee54364c7575f547ec52542a778a65de181a70d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-13 04:25:43.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ca1ee3f5b20cf744f28baab26fa3469bd3c56402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 05:36:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb855ae77d1a4ed7972321387c7891c6\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c45872ac25d4ebda09cf3cd85b92f29 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab3d63aa1ddb40c6903ed8a7cb67b4d7 : SUCCESS in 2m 29s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/818af25fb67d4aeeae994e6cbcf1c595 : FAILURE in 12m 25s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55721a67041543f19fc5c9d60ee50415 : FAILURE in 8m 35s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2615941236a642f2bce2f437ca1d4401 : FAILURE in 7m 47s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d830a0b02917440b98e2fe3f161f747a : FAILURE in 32m 32s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fff5dd8964944c3eba76b675aece830b : SUCCESS in 53m 32s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/a62cad3308e041b0872297c851569af7 : SUCCESS in 18m 55s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5883bab358d418c8c01f2a178df3543 : SUCCESS in 13m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0796fa79af4543b3b370e125b33c021d : SUCCESS in 13m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fc7f75d32394524a6afce117c456e2a : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"85bbdae6150405e1da95de2bc10387517481dd1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-13 07:02:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8f1fa01797de96babcded56950b328a099629b56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 08:04:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c798ef9f91b944d5bffb9e32b43b950d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ee0c5d39c3a422885c9803b4670eb26 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ff58d4bdae145939fb0d6701f79bf13 : SUCCESS in 3m 30s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1b7c5ad625c438aa8151d6be299d7a3 : SUCCESS in 10m 58s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ebf677ec7814d948c32eb5703c49102 : SUCCESS in 7m 25s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83738c61a5494c0daa7e04b148e15375 : SUCCESS in 7m 10s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/906c73dfacc14efcab9dbe44fe6e4ee1 : FAILURE in 41m 20s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/afb43dd98d374ca38e30d043fa9cfc56 : SUCCESS in 55m 54s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/db8fa22bc1c64617894fb3f0ca4e80ad : SUCCESS in 17m 30s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59348fa1a1914a6f928ca4dfb6c479b4 : SUCCESS in 14m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f70972d2ee8141bfb8f68d5fd8fdaf0c : SUCCESS in 14m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50a95eb903a84b0caf64f8f893eecd4b : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":3},{"id":"20508fe3294289cd0e6c747599923c1475226a25","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-02-13 15:38:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"75de060c97aa2552999a01a5c7033f9e1461b3b4","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-02-13 17:12:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1efaeacea2b7d2deb2715a1e9d70db9fb542541e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 17:13:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"766da23a8915c02ac10f230846a7220d177b24f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 18:09:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d712286fe8bb447fa0bd94cb12facf30\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26d7a77d9a44486c90f0ba1f8aae2a0a : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/deb696502c1a483dbc69001ef17750b9 : SUCCESS in 2m 36s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb9c56ba6ac44546859b1a000f94e4dc : SUCCESS in 12m 15s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43cf9bdd9a4140d6a2806033f88da6c1 : SUCCESS in 8m 12s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af059025db8c480298d247ab3c5a8b61 : SUCCESS in 50m 59s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/1544bd4960144b95b2bc13bccff5d1a4 : SUCCESS in 18m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f70c978fe494f22b8f5c0d2d04c553c : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":3},{"id":"a3f031b7b7c99694288d40d9f97fe7182f1ab87e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 18:09:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"edb389b20d7c58a68abf911f63eca06225cb4ff9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 18:10:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bfe1254bd824398ade140207f39a20f\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca013bcffe6e4139b0bf6f5bcdb6a0f8 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0d31bdbdaff4839ea7a19a86f3047b40e4932ebe","revisions":{"7c56aa7efb50393938370387c49e32d619030d4b":{"kind":"REWORK","_number":1,"created":"2023-02-13 03:18:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/873479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/79/873479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/79/873479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/79/873479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/79/873479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/79/873479/1"}}},"commit":{"parents":[{"commit":"66890b154cf4866066906ef6c90cc36461ac8b44","subject":"Merge \"Replace mocha by rspec-mocks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/66890b154cf4866066906ef6c90cc36461ac8b44"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-13 03:16:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-13 03:18:07.000000000","tz":540},"subject":"Remove deprecated manila::service_instance","message":"Remove deprecated manila::service_instance\n\nThis defined type was deprecated by [1].\n\n[1] 8005513f41d43037c92dd0d206df8f41cc66fec2\n\nChange-Id: I8d71ac0ab9fa4a97bc1f5df9a274209198183ddb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/7c56aa7efb50393938370387c49e32d619030d4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/7c56aa7efb50393938370387c49e32d619030d4b"}]},"branch":"refs/heads/master"},"f014a063d0b491e847c8d3a596cf1a9cfbd1314d":{"kind":"NO_CHANGE","_number":2,"created":"2023-02-13 04:25:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/873479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/79/873479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/79/873479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/79/873479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/79/873479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/79/873479/2"}}},"commit":{"parents":[{"commit":"66890b154cf4866066906ef6c90cc36461ac8b44","subject":"Merge \"Replace mocha by rspec-mocks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/66890b154cf4866066906ef6c90cc36461ac8b44"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-13 04:25:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-13 04:25:34.000000000","tz":540},"subject":"Remove deprecated manila::service_instance","message":"Remove deprecated manila::service_instance\n\nThis defined type was deprecated by [1].\n\n[1] 8005513f41d43037c92dd0d206df8f41cc66fec2\n\nChange-Id: I8d71ac0ab9fa4a97bc1f5df9a274209198183ddb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f014a063d0b491e847c8d3a596cf1a9cfbd1314d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f014a063d0b491e847c8d3a596cf1a9cfbd1314d"}]},"branch":"refs/heads/master"},"0d31bdbdaff4839ea7a19a86f3047b40e4932ebe":{"kind":"REWORK","_number":3,"created":"2023-02-13 07:02:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/873479/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/79/873479/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/79/873479/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/79/873479/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/79/873479/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/79/873479/3"}}},"commit":{"parents":[{"commit":"66890b154cf4866066906ef6c90cc36461ac8b44","subject":"Merge \"Replace mocha by rspec-mocks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/66890b154cf4866066906ef6c90cc36461ac8b44"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-13 04:25:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-13 07:02:37.000000000","tz":540},"subject":"Remove deprecated manila::service_instance","message":"Remove deprecated manila::service_instance\n\nThis defined type was deprecated by [1].\n\n[1] 8005513f41d43037c92dd0d206df8f41cc66fec2\n\nChange-Id: I8d71ac0ab9fa4a97bc1f5df9a274209198183ddb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/0d31bdbdaff4839ea7a19a86f3047b40e4932ebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/0d31bdbdaff4839ea7a19a86f3047b40e4932ebe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[]}
