)]}'
{"id":"openstack%2Fpuppet-manila~877239","triplet_id":"openstack%2Fpuppet-manila~master~I28ba3ff1804207c5f93bf3f5352c4257019b36df","project":"openstack/puppet-manila","branch":"master","topic":"cron-ensure","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-13 16:58:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I28ba3ff1804207c5f93bf3f5352c4257019b36df","subject":"Allow customizing status of cron job","status":"MERGED","created":"2023-03-13 10:34:12.000000000","updated":"2023-03-13 16:59:30.000000000","submitted":"2023-03-13 16:58:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877239-cron-ensure","meta_rev_id":"22e03cbe9c0fd03facb17051d1ca8d94dfa8562e","_number":877239,"virtual_id_number":877239,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-13 16:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2023-03-13 13:41:09.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2023-03-13 16:06:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2023-03-13 16:06:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-13 11:32:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-13 13:41:09.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2023-03-13 16:06:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"eb792d52830e67d0d8e76dd0109d68f7f719ec7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-13 10:34:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edbcbc3a4c28b4fda2961e76766f86e82cf6ad59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 11:32:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e688174eab748c183633c8c908cc7f3\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27d01e0cee0c429590784e774e02de45 : SUCCESS in 5m 04s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ca94fb6e2c9412e9ffab65304c696c9 : SUCCESS in 4m 47s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/044ebbb1afe94ccb940e93b6bd54998f : FAILURE in 10m 59s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5026e615e80846ea9328f0c0b19ed4a7 : FAILURE in 7m 44s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44939024564b49b7b00408bd9a241af6 : FAILURE in 5m 09s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14b4f43d0cb54dce8d21d45a21bf5b2a : FAILURE in 43m 24s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/83ecd569687742b9b293a04d55dd6384 : SUCCESS in 51m 44s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/5da0251e42604e85b08a089336db816e : SUCCESS in 20m 09s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f0b79ee628b4de08cdcd07112cb1cc6 : SUCCESS in 16m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98a2584e5e13415680210af30234eae0 : SUCCESS in 16m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53cf3093ed484cfd9f9ecf0502352eb1 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"46b5ab42d7e65be21902b9df95e64a37a80a2a4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-13 11:39:05.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"809c47b91710382cf075dbe0b5c9db7e7da8ac64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 12:38:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac0ac3f4514b4ed49dbe8c5e1c427842\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/488e0ecdb9eb4ba9b47f6574c6a425ec : SUCCESS in 3m 08s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d46a11baacb341848ce7da0552dfd938 : SUCCESS in 2m 15s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c35417ae7aa4f4a950a2bcaff6a6873 : SUCCESS in 11m 41s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/534f05f84c0c4d5d83b86979fa14c316 : SUCCESS in 7m 58s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76a271f0a1644cddba39dd895498a894 : SUCCESS in 6m 57s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f04903cc0a648fabed74eb0cc2f209a : FAILURE in 43m 24s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9e3d8db099a453598dff58cc979411f : SUCCESS in 56m 48s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d32137c4c24d456ba3dad9d935759fa2 : SUCCESS in 18m 39s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd25ab96bcf2477ba36e719fb06860fa : SUCCESS in 15m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ccdd036f6094179828153acf19d1df3 : SUCCESS in 12m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cdf2860b20445e6a787b66230442a9e : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":2},{"id":"a332f5298439ddae4ed6411d14c5cd20b1904e37","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-03-13 13:41:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4a5cb0fa38c8cdae48e9fc124ee165c544b61e9c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-13 16:06:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"653d718cb47b96a9cd6ea67612d2a53f1e80d225","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 16:07:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"548fb92d79d375409b6c6d58904d313d4cdbca4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 16:58:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57eb30ce4a664a36ba14f2ba6856c8e2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ebd201de46c467faddbcca104938c29 : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cafc793cb0b0426e84551b3a842bbf6d : SUCCESS in 2m 40s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e830709525d452285968a50611817c2 : SUCCESS in 11m 18s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d25d652dd83145a7b761117eff42958d : SUCCESS in 8m 24s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/435b0b27f5604242850f474db948b9c6 : SUCCESS in 49m 27s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a2cf4c84fde4e90b57449b78bc8df12 : SUCCESS in 19m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1507b8f6d3ed42819be676d00d41af09 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"aa0ed92eaafe4bf50434b3a2d50808a9eb7d3775","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 16:58:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"22e03cbe9c0fd03facb17051d1ca8d94dfa8562e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 16:59:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/008757b2116d4d4e895729ca0a90b1a2\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1999593d01674f3bbe40a0115df4d6d9 : SUCCESS in 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9dd4512cacdaaa627e4f285586311bdb914396e","revisions":{"0c80e7d30c99b5f31f812e29ab283dc6a7fa4609":{"kind":"REWORK","_number":1,"created":"2023-03-13 10:34:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/877239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/39/877239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/39/877239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/39/877239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/39/877239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/39/877239/1"}}},"commit":{"parents":[{"commit":"1ef4ab6ebde89e8db39f30b498a2535ed0a866fd","subject":"Merge \"Always configure [DEFAULT] auth_strategy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1ef4ab6ebde89e8db39f30b498a2535ed0a866fd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-13 10:33:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-13 10:34:02.000000000","tz":540},"subject":"Allow customizing status of cron job","message":"Allow customizing status of cron job\n\n... so that users can even remove the job. This follows what was added\nto puppet-nova by [1].\n\n[1] https://review.opendev.org/c/openstack/puppet-nova/+/877096\n\nChange-Id: I28ba3ff1804207c5f93bf3f5352c4257019b36df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/0c80e7d30c99b5f31f812e29ab283dc6a7fa4609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/0c80e7d30c99b5f31f812e29ab283dc6a7fa4609"}]},"branch":"refs/heads/master"},"d9dd4512cacdaaa627e4f285586311bdb914396e":{"kind":"REWORK","_number":2,"created":"2023-03-13 11:39:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/877239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/39/877239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/39/877239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/39/877239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/39/877239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/39/877239/2"}}},"commit":{"parents":[{"commit":"1ef4ab6ebde89e8db39f30b498a2535ed0a866fd","subject":"Merge \"Always configure [DEFAULT] auth_strategy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1ef4ab6ebde89e8db39f30b498a2535ed0a866fd"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-13 10:33:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-13 11:39:05.000000000","tz":0},"subject":"Allow customizing status of cron job","message":"Allow customizing status of cron job\n\n... so that users can even remove the job. This follows what was added\nto puppet-nova by [1].\n\n[1] https://review.opendev.org/c/openstack/puppet-nova/+/877096\n\nChange-Id: I28ba3ff1804207c5f93bf3f5352c4257019b36df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/d9dd4512cacdaaa627e4f285586311bdb914396e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/d9dd4512cacdaaa627e4f285586311bdb914396e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
