)]}'
{"id":"openstack%2Fpuppet-manila~892250","triplet_id":"openstack%2Fpuppet-manila~master~I0e39855aa5e4892f338d5f476be7760ec59e2eb8","project":"openstack/puppet-manila","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-08-24 15:05:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0e39855aa5e4892f338d5f476be7760ec59e2eb8","subject":"Ensure absolute path for file creation","status":"MERGED","created":"2023-08-22 01:22:30.000000000","updated":"2023-08-24 15:05:32.000000000","submitted":"2023-08-24 15:05:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892250","meta_rev_id":"6f9f2732c27c584966b04654c16c2be3b543a133","_number":892250,"virtual_id_number":892250,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-24 15:05:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2023-08-22 17:28:10.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2023-08-24 14:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":1,"date":"2023-08-24 14:08:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-22 02:25:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-22 17:28:10.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2023-08-24 14:08:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"906741e900ff62c8afec445bdc9dcbd733a248a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-22 01:22:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5df61becabfe62212f09e5d2a038ae76d8b00677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 02:25:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3e771ccdf9e457eba7c913702386a13\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1e18d16801b41e3865992d3d33506e4 : SUCCESS in 2m 24s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39eb6a07a28942699b58c8a5417a7404 : SUCCESS in 2m 17s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/8760801e230047a982de4d524003f052 : SUCCESS in 5m 34s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8fe9dbcadfb41829e85d3dced96caab : FAILURE in 10m 26s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0b4098423c84f4e90f581b9b28c4c51 : FAILURE in 7m 34s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/221ba3b031fb483fa629cc20b56a8adf : FAILURE in 12m 04s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2eebf66e1c74f0a9938ca7d80629272 : FAILURE in 7m 37s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aace705b43ee45d0ad62d505b0f9a2a5 : SUCCESS in 41m 53s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f45e1ac51e2a4de5a55b1cedb2594a83 : SUCCESS in 57m 51s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1191a4aa49264449bdc2f4c2508878ff : FAILURE in 8m 12s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a657843f632a41e2b4b44643812149b9 : FAILURE in 8m 52s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6c2de06de804d4ab2f9cf7f54e8af2f : FAILURE in 10m 44s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a630b2f4d4ce49589d3b115c07747fe8 : FAILURE in 5m 25s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8edeb866cf44c298159c17d38fbcc8e : FAILURE in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42cf326ab3a8968907434ffa3c0f58485ba72220","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-22 02:58:21.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"50a31e2e097316762389cccefc7693133fb6c25a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 04:06:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/898e182b95a6485e805ed5367161d86e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74d5cc59a70849fcab914bd9ccb7c836 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c7bb3c594d34401929df84ee0f96b2d : SUCCESS in 3m 07s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/9429f7e85ef54b54a0d6becb4885c754 : SUCCESS in 6m 00s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2fc1dd9fc9c49d4b032c2f3fc634611 : SUCCESS in 11m 42s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96eb86c2ca1e4a95b54af9ddb55dc8b6 : SUCCESS in 12m 46s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/81a7c6bea7cd4563bd057950f9ef664e : SUCCESS in 13m 54s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8b7c976baa0477d9508e28bc1eacce5 : SUCCESS in 11m 53s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b7fd00e21e14c2a8bc806bd6972c57b : SUCCESS in 42m 16s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0431e556430d4e00913b1f8b426d99bc : SUCCESS in 57m 54s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f15a715f493e47788a0c199ac5c7bca9 : SUCCESS in 43m 52s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/345df0a93464414eb432ec7d9a582d00 : SUCCESS in 1h 02m 20s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbcac9f9cde14cff8d7e9866b49c5503 : FAILURE in 28m 26s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68f5d3771fb64c8b92286ba4ff70147d : FAILURE in 27m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6c8cbdd865544e48c71cb4b495567c4 : FAILURE in 26m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bdccf99b849befe6ba5043d04adb6dbb4479e275","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-22 12:56:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"527bfc448708d2fd17588cbc5022291a399b20d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 14:02:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa39657f799845a580a85105453f2cf6\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f99e07f5db4435696a8c18e01ebc61e : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6872ddfa79404f2e88e342ffe194b3a3 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b705a3c800b14bafa8d22a8c9454c099 : SUCCESS in 5m 38s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/87dfb6b628584a559e246f75916ceefd : SUCCESS in 11m 40s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/933ab21455ee4dec97c886ce13a5b149 : SUCCESS in 9m 58s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/62804abf491f4d22ad0f06bc92f44326 : SUCCESS in 11m 48s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eac6b7a7b81740fc8e8a6bf12cae87b8 : SUCCESS in 9m 46s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ca60d0af0234d559290b67ae1a87ce7 : SUCCESS in 42m 33s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/132d64ed27db4e3987cb21067af5a5d4 : SUCCESS in 1h 00m 12s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca17d342e49c4d1095f35e8def430af6 : SUCCESS in 49m 06s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e561477b3dc421bac68cf74718fb882 : SUCCESS in 58m 52s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1db0d838a154b378436007ca2e4c2f1 : FAILURE in 29m 19s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cd90a4e2c8047f896b287626d93ebf7 : FAILURE in 24m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/999211806c5b4cb09723b72836a80f8a : FAILURE in 24m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0cee683d4df2cea3a1acef9df9030606457c83a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-22 14:29:12.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b4eac46cd6288e5be964283499be0a66cf39bb6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 16:00:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb676025fc844a6daa6afe2ba9a315b4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8dda62d40b9e4a65b0651c2182109c45 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76689ee56703400ea6795c8bfe6087cd : SUCCESS in 3m 07s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0309ab719c649068f3df07deb152cbd : SUCCESS in 5m 22s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb261f3651c840de9771fbd08ba90e7d : SUCCESS in 11m 23s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0a8240690f64410aa52b1f7e6b393b1 : SUCCESS in 10m 01s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/e045bc8d02724a4b8f01b31667133562 : SUCCESS in 12m 16s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4235afdb1c34db18648e9a15d0741dc : SUCCESS in 7m 35s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/749e902a135e4dd9bfeb140d722e8fc8 : SUCCESS in 43m 24s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cef5311d86b489bbb84faf4094976f2 : SUCCESS in 52m 20s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49084efe997948149cb7a2afb97122c1 : SUCCESS in 38m 29s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec4f057ea48a4ca789dbf9b698966686 : SUCCESS in 53m 27s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b88da14a7b84bb382391204eb170967 : SUCCESS in 16m 54s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73f2f9f96b084417ba005768294b17ba : SUCCESS in 14m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4895074a662c453886684268ea15520e : SUCCESS in 16m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1b44b18fda05f21746c4cd0bed4993ec7d4d8dd9","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-08-22 17:28:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f1db9ba6166e9da95066bfa0fb2e97a9045ed3b1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-08-24 14:08:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0152201c489ed2bafcb8fa5a9d0cf46df6e66305","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 14:09:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a6ff6982c31e2b9bd87235393192cfb7f5e867d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 15:05:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c25621dd760453e9aabc3f33f904a43\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff77cf54369c411f874c180b03fafde0 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8135d0ad5bd148de81e076950f4f9fe1 : SUCCESS in 3m 04s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/94ca1aa1879f4699b5d4088ba5134da8 : SUCCESS in 11m 48s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f083c6a443514039a8c2bb3d016f6c37 : SUCCESS in 9m 44s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/f357adf2148c4cd58f407902cb976318 : SUCCESS in 11m 20s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5833383b089b4f7381be3af23a434d95 : SUCCESS in 46m 28s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cc5dcda29df47888ad7afbca2d44b56 : SUCCESS in 55m 11s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/de38d51fa6bc4f79b7e0d83e96a41143 : SUCCESS in 20m 47s","accounts_in_message":[],"_revision_number":3},{"id":"6f9f2732c27c584966b04654c16c2be3b543a133","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 15:05:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"30c74b8be614989e02ff21a8739dd16af3ad0323","revisions":{"9436614a00a50c5447fee9d7ddcc0120040468f7":{"kind":"REWORK","_number":1,"created":"2023-08-22 01:22:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/892250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/50/892250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/892250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/892250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/892250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/50/892250/1"}}},"commit":{"parents":[{"commit":"8956637035501feca30f5111560df84f411fa24a","subject":"Add .pdkignore to exclude some files from tar ball","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/8956637035501feca30f5111560df84f411fa24a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-22 01:21:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-22 01:21:32.000000000","tz":540},"subject":"Ensure absolute path for file creation","message":"Ensure absolute path for file creation\n\nThis ensures the parameters used as file paths accept only absolute\npaths, so that more sensible errors are returned to users instead of\nthe error from the file resource.\n\nChange-Id: I0e39855aa5e4892f338d5f476be7760ec59e2eb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9436614a00a50c5447fee9d7ddcc0120040468f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9436614a00a50c5447fee9d7ddcc0120040468f7"}]},"branch":"refs/heads/master"},"702d3af1b98c58010607edd6b948bbd8160ebe02":{"kind":"REWORK","_number":2,"created":"2023-08-22 02:58:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/892250/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/50/892250/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/892250/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/892250/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/892250/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/50/892250/2"}}},"commit":{"parents":[{"commit":"8956637035501feca30f5111560df84f411fa24a","subject":"Add .pdkignore to exclude some files from tar ball","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/8956637035501feca30f5111560df84f411fa24a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-22 01:21:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-22 02:58:20.000000000","tz":0},"subject":"Ensure absolute path for file creation","message":"Ensure absolute path for file creation\n\nThis ensures the parameters used as file paths accept only absolute\npaths, so that more sensible errors are returned to users instead of\nthe error from the file resource.\n\nChange-Id: I0e39855aa5e4892f338d5f476be7760ec59e2eb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/702d3af1b98c58010607edd6b948bbd8160ebe02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/702d3af1b98c58010607edd6b948bbd8160ebe02"}]},"branch":"refs/heads/master"},"30c74b8be614989e02ff21a8739dd16af3ad0323":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-08-22 14:29:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/892250/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/50/892250/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/892250/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/892250/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/50/892250/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/50/892250/3"}}},"commit":{"parents":[{"commit":"7389c58a17324ed8c207b2d9a64abc13c65422f9","subject":"Enable memcached in acceptance tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/7389c58a17324ed8c207b2d9a64abc13c65422f9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-22 01:21:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-22 14:29:12.000000000","tz":0},"subject":"Ensure absolute path for file creation","message":"Ensure absolute path for file creation\n\nThis ensures the parameters used as file paths accept only absolute\npaths, so that more sensible errors are returned to users instead of\nthe error from the file resource.\n\nChange-Id: I0e39855aa5e4892f338d5f476be7760ec59e2eb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/30c74b8be614989e02ff21a8739dd16af3ad0323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/30c74b8be614989e02ff21a8739dd16af3ad0323"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
