)]}'
{"id":"openstack%2Fpuppet-manila~897510","triplet_id":"openstack%2Fpuppet-manila~master~Idf187458b62e81999874f9582f932d9707c0d8c1","project":"openstack/puppet-manila","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-10-11 04:48:26.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2023-10-09 14:45:26.000000000","reason":"\u003cGERRIT_ACCOUNT_21129\u003e replied on the change","reason_account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}},"hashtags":[],"change_id":"Idf187458b62e81999874f9582f932d9707c0d8c1","subject":"Add support for share backup/restore options","status":"MERGED","created":"2023-10-06 03:31:09.000000000","updated":"2023-10-11 07:00:05.000000000","submitted":"2023-10-11 06:58:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897510","meta_rev_id":"7891c06f4cc7b4ae84e94424a7240476113ab0ef","_number":897510,"virtual_id_number":897510,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 06:58:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-10-09 05:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-10-09 21:43:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-10-09 21:43:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-06 06:40:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-06 16:40:50.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2023-10-09 05:53:00.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-10-09 21:43:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"1cdcb5c94f2443a5b971c232ad6fd4f1fe5e5bc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-06 03:31:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89a4161300a13f592a15e5f8288fb8372cb2ac33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-06 04:31:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f09ed787249dad0b752d903aa328b08f6fc3bcee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-06 05:06:20.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"847c6e14a7fe0e7d668f807a1698c467b03c5ca1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-06 05:20:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1d90f95dc11c25e606035876996414c750102946","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-06 05:37:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"09ed0fbe781e8ec21249b18d33985b1cfd64324e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 06:40:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/101bd9f0fd4d4b6b9112c163dcbc88bc\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2b3393f152c480faa83e8a2adc3d638 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2556da81058b41c0aa405454474e26d5 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4760a98918fa45eaa5a5addff723329a : SUCCESS in 4m 30s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/414d69a4363040d39315f0668b729867 : SUCCESS in 13m 38s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0561184df53d46dc835027c8f4a90ad5 : SUCCESS in 7m 59s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b096b66842d4a98a357fac498fcda85 : SUCCESS in 9m 36s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7095e29d2044ffa9ed8b8064435f043 : SUCCESS in 8m 56s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4cc789d04f8f4653b0993d6cf714780d : FAILURE in 43m 55s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ceb24cccf0f64e8bb6401501a3fd3920 : SUCCESS in 55m 58s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a71c539016f04a47b0a25345ed31907b : FAILURE in 41m 09s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d92a9725ae7644ed9575c76fbb369fd9 : SUCCESS in 57m 59s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/944e991010fe48e0b04b492990f2f9a6 : SUCCESS in 16m 23s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e6e277fb8de4b20aed28b4dbb4ec860 : SUCCESS in 14m 50s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de69aedc48f14bdf991df11c727adf5a : SUCCESS in 11m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59bf124767144a9ab8cddefc6174d889 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":5},{"id":"1b59502278d6434393e7d588b44efda8cebf01c2","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-10-06 16:40:50.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"01a0d66e32d97a32449611828fe0c08e01213fd6","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-10-06 16:49:51.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1dc65cc5ec8ad7e5fe13c32f69e6eba7d3be311d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-07 02:20:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"24c262fde48bfeb44e8bc013aac26765a69d9a8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-07 02:30:38.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"bbfbb13c02f36d5d99ed10c2ebef5c6bb03c0681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 03:46:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b094a65e7cd148919e25bb6968b384c5\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3200f298f742472099420b2db9501fe4 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3af19805a7314e16a4b399ab298babc8 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b695072fac054396b8a66a1c0df1e476 : SUCCESS in 6m 26s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3aa68d915854ce0b3dd7e681b9c7d2d : FAILURE in 15m 37s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e43adb7c36bf4d0588f1263c358ca959 : FAILURE in 9m 17s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/19e8050640214a9d8b73f5477563a708 : FAILURE in 12m 05s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db52d02aa3794b48a458c9340afedf10 : FAILURE in 7m 45s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cece364173024311ae48ff6b5e9982a9 : FAILURE in 46m 40s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fcb73df00ffd441da5f2efc2b7333ade : SUCCESS in 1h 03m 56s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4a6e7ee07a84d91a1cf9039a6041f2e : FAILURE in 51m 24s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a63a7a905ce44bc6b5a29b754b3f05ce : SUCCESS in 57m 41s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/978b5bff81a04cd1b5ff6d12ea1e627e : SUCCESS in 20m 41s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1f74879303a4a4dbd6fc12c067d92a0 : SUCCESS in 15m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c8b7a31f92d427fa55e23aa82a86db9 : SUCCESS in 14m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f498a23fc49c4bb49b060798ca53a8f0 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":6},{"id":"297b1a511db8314d07242e97cb607ab7e11b73d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-07 08:03:37.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"8c7c0323963bac157aea8c29982b57fcf338a950","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 09:09:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/497098c527c04253a5d00ca322a20463\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/297fff5358ee4da7abf8c7c19a56a9b1 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2439beb5dd3d4a19bdcf1ab029dd1ef1 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bb3639b445e486d9c61e626c64c7a7c : SUCCESS in 5m 53s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/07f19f3c5ef54fc4aabc4ba31b159d90 : SUCCESS in 10m 14s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2987690d57b142de853c7de3717d709f : SUCCESS in 7m 28s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f2d3b4d7a3c461198203aa9b19283e6 : SUCCESS in 11m 22s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b137de621524a3a9abf293e5053ccf0 : SUCCESS in 7m 06s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1749eae2abc34939a1d4c46b5a4cad42 : FAILURE in 41m 49s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf1515a15a444adc9f3fa37a22f73d8b : SUCCESS in 59m 06s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d91fe273c9ba44388779c6d287cf2689 : FAILURE in 45m 01s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4898d3f6bd06448ca97833d5945d0cbd : SUCCESS in 53m 07s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e1977d462b94d6c8ff5691a79ef3525 : SUCCESS in 17m 58s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c28f8fa4ef14075b01254a173d841df : SUCCESS in 13m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65f745c23f0e446bab79112b2082a2b3 : SUCCESS in 13m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80a8568ca0be409d8a10eddc3304a448 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":7},{"id":"3a628c6c1ccad293cd1de9b934a81420bffdc634","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-10-09 05:53:00.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"c9f28784656d7f145573fcce3f676155f11531a0","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-10-09 14:45:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0ab70119ae5d027dc14cd870ee8c7939701b12d2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-10-09 21:43:50.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"4b5f37e796cc661ee808d3ee8350fb421fff7902","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 21:44:42.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"451670dd76936704687e39b8f0a3766569652964","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 22:51:11.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/576ac5d89e8e46ff8e1171a873f42be6\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdacdf845b174c8dac7774b920fd2fc4 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b9f0ab038e64b668bc3c3ebfa431716 : SUCCESS in 2m 30s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/e63f5a20a86643f9aa5d13eeaa1bdf4b : FAILURE in 7m 49s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b6158fca0854ed4b4731990a4f74c48 : FAILURE in 6m 34s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/f82feb845ec541dfaf402348f5fe5b7a : FAILURE in 7m 57s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7037c8565e2c43ba9ce4a642a5f9b38e : SUCCESS in 1h 00m 13s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa108a7e66e0414ba1dce3cc7b3377e9 : SUCCESS in 55m 22s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/796108e797054e889f2e8c47ed520905 : SUCCESS in 20m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a5ed75efde540b797bb3c34ad83d1e3 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":7},{"id":"f402b1c48748047d01d4ddba270499924fbd304c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-11 04:48:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"92f515548d0988009d964c190b786562f867d716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 06:08:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43e342f03c3a42e0bdd681f04af61420\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da74b4602c764fc4a5aa09ac2736a24a : SUCCESS in 2m 55s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/015467a34eca41a5a43f00d67d3221f7 : SUCCESS in 2m 56s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/45329e4d13c1459f961e37057077c23b : SUCCESS in 5m 59s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/60c5afd1275843438cc2ece0c428451d : SUCCESS in 11m 09s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1abcd5cd4044823b44aaf1526657fba : SUCCESS in 9m 11s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec14018d8d6f435fbf0a81d819af61ec : SUCCESS in 14m 03s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d05a15e0f580439ea17468b329a0e3a0 : SUCCESS in 7m 44s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6ee867382d240728678dda4d9bb7944 : FAILURE in 44m 22s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/84c389f04445465c8386ccdc00985526 : SUCCESS in 1h 17m 26s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ca78c652b094e4ab1aa5ecc095f2b5d : FAILURE in 45m 56s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9f0f203899874353806ac9b93f805cd7 : SUCCESS in 52m 47s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/52f3ca5f463e430583165d0342de539d : SUCCESS in 19m 55s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60a05b87bc8c4e9f99fb853739b2d185 : SUCCESS in 16m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62c7b91490e64b7eac277ed755b9621f : SUCCESS in 15m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d91512ee402d4a989e7ca315c38954ef : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":7},{"id":"50aac891fff1e990e032e704bf9eba0f973fadaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 06:09:11.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f7e332d7e001881c4efa1b405400cb37b9ca6143","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 06:58:47.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b870e5affdfe41e3878600d52b44c50f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd05b0b48db8435b8664dc480178045c : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04e361fab165435baca98d30e35cdfb6 : SUCCESS in 2m 44s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e7a68e629664c2bac8c58d9443a171f : SUCCESS in 11m 30s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b71874cd5ddf4aaa8a4d97b5fdc408d4 : SUCCESS in 8m 31s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/870bffabc5894d6dacbd080fb0e3f144 : SUCCESS in 11m 36s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b258a7beeb8e44f193e8e3a791a72d09 : SUCCESS in 49m 11s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c22d5cda7ed84e5d81ede205201739ae : SUCCESS in 41m 58s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/dee4ed60916c46de8c975f70a9778825 : SUCCESS in 20m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a658c4263c44261bbcd12b967dfaba6 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":7},{"id":"2e6569370e53daa179167b42792d33783c60da47","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 06:58:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"7891c06f4cc7b4ae84e94424a7240476113ab0ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 07:00:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b63a3ecca2c4f0f862059deb173c9cd\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b26c08e1e67411894a2db130bd66225 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2c6e685768ff23a92350c2cc999c3b7d48914464","revisions":{"39562097c352d15c0915201cee1f3ca11cd0c769":{"kind":"REWORK","_number":1,"created":"2023-10-06 03:31:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/897510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/10/897510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/1"}}},"commit":{"parents":[{"commit":"f779495bd22ec859272a86a414ac5652e12073d6","subject":"Merge \"Prepare 2023.2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f779495bd22ec859272a86a414ac5652e12073d6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-06 03:30:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-06 03:30:55.000000000","tz":540},"subject":"Add support for share backup/restore options","message":"Add support for share backup/restore options\n\nChange-Id: Idf187458b62e81999874f9582f932d9707c0d8c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/39562097c352d15c0915201cee1f3ca11cd0c769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/39562097c352d15c0915201cee1f3ca11cd0c769"}]},"branch":"refs/heads/master"},"dc42c64a90fff569dcd54874856219b9771fc147":{"kind":"REWORK","_number":2,"created":"2023-10-06 04:31:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/897510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/10/897510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/2"}}},"commit":{"parents":[{"commit":"503f4cd68fb22cff0e9ac389fc6ceff1852650f1","subject":"Add support for share manger options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/503f4cd68fb22cff0e9ac389fc6ceff1852650f1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-06 03:30:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-06 04:29:01.000000000","tz":540},"subject":"Add support for share backup/restore options","message":"Add support for share backup/restore options\n\nThis introduces support for options used by share backup and restore\nfeature, which was introduced to manila recently.\n\nChange-Id: Idf187458b62e81999874f9582f932d9707c0d8c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/dc42c64a90fff569dcd54874856219b9771fc147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/dc42c64a90fff569dcd54874856219b9771fc147"}]},"branch":"refs/heads/master"},"1cca658b59aa62f5f2b102dafd1c54d91c9bc8c6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-10-06 05:06:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/897510/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/10/897510/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/3"}}},"commit":{"parents":[{"commit":"f319ebb1459a38073a67bf383fa537763dce77fb","subject":"Add support for share manger options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f319ebb1459a38073a67bf383fa537763dce77fb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-06 03:30:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-06 05:06:02.000000000","tz":540},"subject":"Add support for share backup/restore options","message":"Add support for share backup/restore options\n\nThis introduces support for options used by share backup and restore\nfeature, which was introduced to manila recently.\n\nChange-Id: Idf187458b62e81999874f9582f932d9707c0d8c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1cca658b59aa62f5f2b102dafd1c54d91c9bc8c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1cca658b59aa62f5f2b102dafd1c54d91c9bc8c6"}]},"branch":"refs/heads/master"},"e7d60ddda20ee71dc16dd3f02afd1a41883bf51f":{"kind":"REWORK","_number":4,"created":"2023-10-06 05:20:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/897510/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/10/897510/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/4"}}},"commit":{"parents":[{"commit":"f319ebb1459a38073a67bf383fa537763dce77fb","subject":"Add support for share manger options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f319ebb1459a38073a67bf383fa537763dce77fb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-06 03:30:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-06 05:20:06.000000000","tz":540},"subject":"Add support for share backup/restore options","message":"Add support for share backup/restore options\n\nThis introduces support for options used by share backup and restore\nfeature, which was introduced to manila recently.\n\nChange-Id: Idf187458b62e81999874f9582f932d9707c0d8c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/e7d60ddda20ee71dc16dd3f02afd1a41883bf51f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/e7d60ddda20ee71dc16dd3f02afd1a41883bf51f"}]},"branch":"refs/heads/master"},"8478cc1ecb1ad18f6674586531aef7225706c6c0":{"kind":"REWORK","_number":5,"created":"2023-10-06 05:37:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/897510/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/10/897510/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/5"}}},"commit":{"parents":[{"commit":"f319ebb1459a38073a67bf383fa537763dce77fb","subject":"Add support for share manger options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f319ebb1459a38073a67bf383fa537763dce77fb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-06 03:30:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-06 05:37:31.000000000","tz":540},"subject":"Add support for share backup/restore options","message":"Add support for share backup/restore options\n\nThis introduces support for options used by share backup and restore\nfeature, which was introduced to manila recently.\n\nChange-Id: Idf187458b62e81999874f9582f932d9707c0d8c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/8478cc1ecb1ad18f6674586531aef7225706c6c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/8478cc1ecb1ad18f6674586531aef7225706c6c0"}]},"branch":"refs/heads/master"},"3008c6ad3eb15cfe87dc8a660ee9155a638c053f":{"kind":"REWORK","_number":6,"created":"2023-10-07 02:30:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/897510/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/10/897510/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/6"}}},"commit":{"parents":[{"commit":"f319ebb1459a38073a67bf383fa537763dce77fb","subject":"Add support for share manger options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f319ebb1459a38073a67bf383fa537763dce77fb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-06 03:30:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-07 02:27:45.000000000","tz":540},"subject":"Add support for share backup/restore options","message":"Add support for share backup/restore options\n\nThis introduces support for options used by share backup and restore\nfeature, which was introduced to manila recently.\n\nThis also adds support for NFS share backup driver.\n\nChange-Id: Idf187458b62e81999874f9582f932d9707c0d8c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/3008c6ad3eb15cfe87dc8a660ee9155a638c053f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/3008c6ad3eb15cfe87dc8a660ee9155a638c053f"}]},"branch":"refs/heads/master"},"2c6e685768ff23a92350c2cc999c3b7d48914464":{"kind":"REWORK","_number":7,"created":"2023-10-07 08:03:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/897510/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/10/897510/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/10/897510/7"}}},"commit":{"parents":[{"commit":"f319ebb1459a38073a67bf383fa537763dce77fb","subject":"Add support for share manger options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f319ebb1459a38073a67bf383fa537763dce77fb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-06 03:30:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-07 08:02:16.000000000","tz":540},"subject":"Add support for share backup/restore options","message":"Add support for share backup/restore options\n\nThis introduces support for options used by share backup and restore\nfeature, which was introduced to manila recently.\n\nThis also adds support for NFS share backup driver.\n\nChange-Id: Idf187458b62e81999874f9582f932d9707c0d8c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/2c6e685768ff23a92350c2cc999c3b7d48914464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/2c6e685768ff23a92350c2cc999c3b7d48914464"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
