)]}'
{"id":"openstack%2Fpuppet-manila~899441","triplet_id":"openstack%2Fpuppet-manila~master~I56951a77264154c2f3cbc57234b7f54eef99aee1","project":"openstack/puppet-manila","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-10-30 14:44:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I56951a77264154c2f3cbc57234b7f54eef99aee1","subject":"Allow skipping installation of storops library","status":"MERGED","created":"2023-10-27 01:41:41.000000000","updated":"2023-10-30 14:45:00.000000000","submitted":"2023-10-30 14:44:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899441","meta_rev_id":"5d4bbbf700e947c856d144cf7fdebfc49354f682","_number":899441,"virtual_id_number":899441,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-30 14:44:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-10-30 08:05:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-10-30 12:57:51.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-10-30 12:57:51.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-27 02:49:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-30 08:05:03.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-10-30 12:57:51.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"5748484696c9a460396b79f0bd24d7cde2379a98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-27 01:41:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bba3760f210e2c5682d0b4dac53ab2f8a924a83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-27 01:42:54.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"425fd9cff84bdcb965f78279cdeca9aaaa63716e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-27 01:43:34.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d3ee8455ecbec07d190a8b330ec941a2f1395885","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 02:49:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11dcfca3ac724c52a4438157120b1ad8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e956863490544f69dfaac029b602f47 : FAILURE in 2m 32s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b1efa748612438d8cd9e6c1ff0f2861 : FAILURE in 2m 35s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef936093a3864bc18418896c42453a76 : FAILURE in 5m 12s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/4691a7811fe2461cbd009a85756d38c6 : FAILURE in 8m 21s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1357e9cd21714c2d9b76b375e15b70fe : FAILURE in 4m 53s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ca4db4f0e774ac89ec3821898905051 : FAILURE in 7m 30s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93754f6505f54df190e13b99d688e1ce : FAILURE in 5m 57s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c173d2d78c94711973b27428452e152 : FAILURE in 54m 00s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ff60eaa6bf648c281c72cf56f602177 : SUCCESS in 59m 11s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9decddb243d46d58a09fc35ca8445b7 : SUCCESS in 42m 34s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/18150c2b1ab145fdadd36e1301883b55 : SUCCESS in 51m 28s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4b59e22f51e4dafbd81a8d4bc68b519 : SUCCESS in 19m 06s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d9ccd30494f4edfbff6c450c9dbcd9f : SUCCESS in 13m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/541e445930fd40788a53ad1bef46be0c : SUCCESS in 12m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c623bfea2637498aa20e549c11e764c8 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":3},{"id":"4729b61c8992b0dbb3bda0d74907726b71aa05d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-27 02:50:31.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"954cb62acc6fd13ce3cf17b36180734932242b28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 04:17:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74b634d0810b48cba4942e8500315123\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a216fe233014322bfe8e0c153367ba2 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22b73d75862e4cf5a12cdf6d2ab8be62 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b6be74d30714f3e91c64f6f8bdf172c : SUCCESS in 6m 41s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/c62ce675e7d54d4a96c1886181a82dc5 : FAILURE in 9m 29s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a647f0f7cd71418894106f6c5ce3f874 : FAILURE in 6m 03s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/664bfb8a7a884aabbac94d01e71b2b3a : FAILURE in 9m 09s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65fb82cca39042ac90f7a6deed84d5ce : FAILURE in 6m 22s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/609f3ae39ca54fa09ec67b5757489781 : SUCCESS in 47m 31s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e6c4be51c8547be83a6521a78881736 : SUCCESS in 59m 16s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec218bdafe154e34963d0015666936ee : SUCCESS in 41m 37s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/86b6149518324eaf88e39d6376af827c : SUCCESS in 1h 02m 00s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/68efcf1e64a04d95a76c59d8e8a7320e : SUCCESS in 21m 15s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88bbe8641eed4718ba242b5edc4bf3dc : SUCCESS in 16m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10ae04509f4d4f58806adcb14b3d1bfb : SUCCESS in 14m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68d50df04c964528aefe64c02499c7e9 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":4},{"id":"c8b30a4c2a1d69ca77ba6dbc0d7a4ffe2debbfd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-27 11:25:56.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"40075ac3765e5edd2ed2da3c2c265d5091466dcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-27 11:30:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d6b8dea13d3239a6a07f5c9caa54b47bb1ea4d85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-27 13:41:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"118785f5e53c0e788108db409297ed95676dd086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 15:57:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d4e0e461a9743daa766bcceacf62350\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34b2a0c3032547d295f15a910b420342 : SUCCESS in 2m 36s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f65db8832d44e618589b17bbf529295 : SUCCESS in 2m 46s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/e266ff19c9844a6d8adefdab0dca921b : SUCCESS in 5m 04s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/5336687cdb3c4adcaad39d01017fb1ef : SUCCESS in 10m 43s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55702d1c9b3543fc875919b0bf8e2fd7 : SUCCESS in 7m 37s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcf5dbc653c948f4a1e30103a2eb355e : SUCCESS in 13m 43s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4cb1452c80644c9b958b3a71433a152 : SUCCESS in 10m 25s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ad2414cd91a4bbca48f7dd96e9c632a : FAILURE in 49m 30s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7905eaa616734ef59301222d7d1a0042 : TIMED_OUT in 2h 03m 12s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7acfaddd6d6e4a32bf5a59201cdcd7a0 : SUCCESS in 47m 21s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4580ae405221481983b247781d88cb13 : SUCCESS in 1h 00m 49s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/0807fb8e865e452988dd64680cfa2716 : SUCCESS in 17m 37s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d19b69084b649799d95e0cb9b8e2e86 : SUCCESS in 14m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cdb7baf4320148bcbda79ff02d250599 : SUCCESS in 14m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d05d91dfc4d640efa47db71aea1b168c : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":7},{"id":"0936d34c1c6422d42e6452a752871eb4b823322a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-27 21:58:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cf594a48f7d000c866f449b0baf90592ad26a981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 22:47:59.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e848d01cc79468dad26d1b576ed5584\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d7c7dc3591443a8a0cc775150fd3cca : SUCCESS in 2m 34s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66a32e8bd94c4462b3365d837049e366 : SUCCESS in 2m 41s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/12d0946340174c9ebab7254de4aa9f43 : SUCCESS in 5m 50s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d9b8230ed764fb5bef3b027c9de3ba5 : SUCCESS in 12m 11s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fc94fe651e947d490b78f7bdac93d5e : SUCCESS in 9m 52s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b9b82a778aa4be398f6c44d5bc61458 : SUCCESS in 12m 37s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6642b55747ef40c7b265bc6638226f61 : SUCCESS in 9m 24s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d3d6c5bf70f4aa1af08ff67c0e1b383 : FAILURE in 41m 53s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2eafc59ae271458695effd9910bf2d50 : SUCCESS in 47m 41s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f313c58ff6e24d29b09a8b8e4f245087 : SUCCESS in 36m 45s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/915133da46834be983e3773adc66c57b : SUCCESS in 44m 31s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc6b3b31e52448cf9ee1a622563a25d8 : SUCCESS in 15m 44s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2052bb629d9e476fa9126b94aeba1b4a : SUCCESS in 12m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8fc35c793fa42d5a1158e258433d7d5 : SUCCESS in 16m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eedf6071efcc41b99ce506585244fed7 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":7},{"id":"cbce4ecb635e492a873fba8d21ddcbdd4a8effd9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-10-30 08:05:03.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"6cc5185d993df0c10ec0715de04875726d994655","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2023-10-30 12:57:51.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"1307a622e80be9ed3dd5a1378157a104dbf61e7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 12:58:44.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"25bb6e179961e4145862ecb92550bedf0eb00eed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 14:44:01.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d108485dae74a539e6a0924ec109aad\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c80245c8f4ce44d6bea25adf6a73d8dd : SUCCESS in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d68b70f320454378a117fe16276fb64e : SUCCESS in 2m 28s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/14d64fc9b70343e8b3dd541a94644121 : SUCCESS in 10m 02s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/898268037142428ea526ad26a2e75c73 : SUCCESS in 6m 53s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/62cc6b99aac54800b0e2433a71499160 : SUCCESS in 10m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/09372b5af4294999bd6dea3a5f19fe1f : SUCCESS in 59m 29s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3550ea0182b04a87b8eb2ef75da0c24b : SUCCESS in 51m 21s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7f731b333c64256a67fb18da8d49309 : SUCCESS in 18m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f4d0bfedbdc4d468d61c560caf30987 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":7},{"id":"130f1a119a1ea8d4fab77790b3e2309dbfdc060d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 14:44:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"5d4bbbf700e947c856d144cf7fdebfc49354f682","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 14:45:00.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/777c559323c447f8af86fcfa58b35a89\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c8d3354cc664b63aa3a99b58aa05d91 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"cfab2b79d19ddaa9dc9c5380d02071b9b0b46e88","revisions":{"7d31230bb3498fc0ad8abc67ef797cea90022e05":{"kind":"REWORK","_number":1,"created":"2023-10-27 01:41:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/899441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/41/899441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/1"}}},"commit":{"parents":[{"commit":"1b89d78af6254c3e0c45f9af75d4d404f4076505","subject":"Merge \"Allow multiple Unity backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1b89d78af6254c3e0c45f9af75d4d404f4076505"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 01:37:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 01:41:12.000000000","tz":540},"subject":"Allow skipping installation of storops library","message":"Allow skipping installation of storops library\n\nThe unity share driver requires storops library, and now we install it\nusing pip. However this may not be ideal in some deployments because\nit mixes up python libraries installed by packages and ones installed\nby pip. This change introduces an option to disable management of\nthe storops library so that users can install the library b\na different method (eg. using a package they build).\n\nThis change also fixes the missing manila-support-package tag, to\nensure the package is installed before manila services are started.\n\nChange-Id: I56951a77264154c2f3cbc57234b7f54eef99aee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/7d31230bb3498fc0ad8abc67ef797cea90022e05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/7d31230bb3498fc0ad8abc67ef797cea90022e05"}]},"branch":"refs/heads/master"},"729f6cb68b38f1ceb5a06ce30fe7b5abe997e817":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-27 01:42:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/899441/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/41/899441/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/2"}}},"commit":{"parents":[{"commit":"1b89d78af6254c3e0c45f9af75d4d404f4076505","subject":"Merge \"Allow multiple Unity backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1b89d78af6254c3e0c45f9af75d4d404f4076505"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 01:37:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 01:42:23.000000000","tz":540},"subject":"Allow skipping installation of storops library","message":"Allow skipping installation of storops library\n\nThe unity share driver requires storops library, and now we install it\nusing pip. However this may not be ideal in some deployments because\nit mixes up python libraries installed by packages and ones installed\nby pip. This change introduces an option to disable management of\nthe storops library so that users can install the library by\na different method (3rd party package).\n\nThis change also fixes the missing manila-support-package tag, to\nensure the package is installed before manila services are started.\n\nChange-Id: I56951a77264154c2f3cbc57234b7f54eef99aee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/729f6cb68b38f1ceb5a06ce30fe7b5abe997e817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/729f6cb68b38f1ceb5a06ce30fe7b5abe997e817"}]},"branch":"refs/heads/master"},"3b9324fb94593328fac2c57565f97d872c442a89":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-10-27 01:43:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/899441/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/41/899441/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/3"}}},"commit":{"parents":[{"commit":"1b89d78af6254c3e0c45f9af75d4d404f4076505","subject":"Merge \"Allow multiple Unity backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1b89d78af6254c3e0c45f9af75d4d404f4076505"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 01:37:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 01:43:06.000000000","tz":540},"subject":"Allow skipping installation of storops library","message":"Allow skipping installation of storops library\n\nThe unity share driver requires storops library, and now we install it\nusing pip. However this may not be ideal in some deployments because\nit mixes up python libraries installed by packages and ones installed\nby pip. This change introduces an option to disable management of\nthe storops library so that users can install the library by\na different method (3rd party package).\n\nThis change also fixes the missing manila-support-package tag, to\nensure that the package is installed before manila services start.\n\nChange-Id: I56951a77264154c2f3cbc57234b7f54eef99aee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/3b9324fb94593328fac2c57565f97d872c442a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/3b9324fb94593328fac2c57565f97d872c442a89"}]},"branch":"refs/heads/master"},"4957a05562bfc0de6e717f65d25645afd43da655":{"kind":"REWORK","_number":4,"created":"2023-10-27 02:50:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/899441/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/41/899441/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/4"}}},"commit":{"parents":[{"commit":"1b89d78af6254c3e0c45f9af75d4d404f4076505","subject":"Merge \"Allow multiple Unity backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1b89d78af6254c3e0c45f9af75d4d404f4076505"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 01:37:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 02:50:30.000000000","tz":0},"subject":"Allow skipping installation of storops library","message":"Allow skipping installation of storops library\n\nThe unity share driver requires storops library, and now we install it\nusing pip. However this may not be ideal in some deployments because\nit mixes up python libraries installed by packages and ones installed\nby pip. This change introduces an option to disable management of\nthe storops library so that users can install the library by\na different method (3rd party package).\n\nThis change also fixes the missing manila-support-package tag, to\nensure that the package is installed before manila services start.\n\nChange-Id: I56951a77264154c2f3cbc57234b7f54eef99aee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/4957a05562bfc0de6e717f65d25645afd43da655"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/4957a05562bfc0de6e717f65d25645afd43da655"}]},"branch":"refs/heads/master"},"2c3c6c9b45b15caf9a5ba0643b734bd3f03bbbab":{"kind":"REWORK","_number":5,"created":"2023-10-27 11:25:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/899441/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/41/899441/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/5"}}},"commit":{"parents":[{"commit":"1b89d78af6254c3e0c45f9af75d4d404f4076505","subject":"Merge \"Allow multiple Unity backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1b89d78af6254c3e0c45f9af75d4d404f4076505"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 01:37:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 11:25:48.000000000","tz":540},"subject":"Allow skipping installation of storops library","message":"Allow skipping installation of storops library\n\nThe unity share driver requires storops library, and now we install it\nusing pip. However this may not be ideal in some deployments because\nit mixes up python libraries installed by packages and ones installed\nby pip. This change introduces an option to disable management of\nthe storops library so that users can install the library by\na different method (3rd party package).\n\nThis change also fixes the missing manila-support-package tag, to\nensure that the package is installed before manila services start.\n\nChange-Id: I56951a77264154c2f3cbc57234b7f54eef99aee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/2c3c6c9b45b15caf9a5ba0643b734bd3f03bbbab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/2c3c6c9b45b15caf9a5ba0643b734bd3f03bbbab"}]},"branch":"refs/heads/master"},"91a6810dd62b5110017290490ffed92abd1c1bc3":{"kind":"REWORK","_number":6,"created":"2023-10-27 11:30:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/899441/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/41/899441/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/6"}}},"commit":{"parents":[{"commit":"1b89d78af6254c3e0c45f9af75d4d404f4076505","subject":"Merge \"Allow multiple Unity backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1b89d78af6254c3e0c45f9af75d4d404f4076505"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 01:37:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 11:30:36.000000000","tz":540},"subject":"Allow skipping installation of storops library","message":"Allow skipping installation of storops library\n\nThe unity share driver requires storops library, and now we install it\nusing pip. However this may not be ideal in some deployments because\nit mixes up python libraries installed by packages and ones installed\nby pip. This change introduces an option to disable management of\nthe storops library so that users can install the library by\na different method (3rd party package).\n\nThis change also fixes the missing manila-support-package tag, to\nensure that the package is installed before manila services start.\n\nChange-Id: I56951a77264154c2f3cbc57234b7f54eef99aee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/91a6810dd62b5110017290490ffed92abd1c1bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/91a6810dd62b5110017290490ffed92abd1c1bc3"}]},"branch":"refs/heads/master"},"cfab2b79d19ddaa9dc9c5380d02071b9b0b46e88":{"kind":"REWORK","_number":7,"created":"2023-10-27 13:41:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/899441/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/41/899441/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/41/899441/7"}}},"commit":{"parents":[{"commit":"1b89d78af6254c3e0c45f9af75d4d404f4076505","subject":"Merge \"Allow multiple Unity backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/1b89d78af6254c3e0c45f9af75d4d404f4076505"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 01:37:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 13:41:31.000000000","tz":540},"subject":"Allow skipping installation of storops library","message":"Allow skipping installation of storops library\n\nThe unity share driver requires storops library, and now we install it\nusing pip. However this may not be ideal in some deployments because\nit mixes up python libraries installed by packages and ones installed\nby pip. This change introduces an option to disable management of\nthe storops library so that users can install the library by\na different method (3rd party package).\n\nThis change also fixes the missing manila-support-package tag, to\nensure that the package is installed before manila services start.\n\nChange-Id: I56951a77264154c2f3cbc57234b7f54eef99aee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/cfab2b79d19ddaa9dc9c5380d02071b9b0b46e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/cfab2b79d19ddaa9dc9c5380d02071b9b0b46e88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
