)]}'
{"id":"openstack%2Fpuppet-manila~953046","triplet_id":"openstack%2Fpuppet-manila~master~I1063eae7d2ef11da6f1bda2c93e6877e8e8e1d2f","project":"openstack/puppet-manila","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-24 01:08:07.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I1063eae7d2ef11da6f1bda2c93e6877e8e8e1d2f","subject":"Remove management of manila v2 user","status":"MERGED","created":"2025-06-22 08:40:49.000000000","updated":"2025-06-24 02:28:42.000000000","submitted":"2025-06-24 02:27:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953046","meta_rev_id":"a476d8c3e6b0c6b7c33be7abc84b6c83b2089761","_number":953046,"virtual_id_number":953046,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-24 02:27:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":2,"date":"2025-06-22 22:18:23.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":2,"date":"2025-06-23 06:48:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-06-23 20:44:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-06-23 20:44:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-22 09:04:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-22 22:18:23.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2025-06-23 06:48:25.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-06-23 20:44:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"ac39086c837d7117b4be20a293908aa783f1a049","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-22 08:40:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb205605df828bb26bda1a97a8d87ddcb0cf1fc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-22 08:44:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"967add72c25b852ab545b9c8b769a5a56a938249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-22 09:04:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd6d5f992db94e9588cf3fde8f54fd74\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94fd2f335fa2419aa362b26ffaa2d008 : SUCCESS in 2m 12s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/838846c09def4e3ab0a76a2b1d05cdce : SUCCESS in 2m 15s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d056b01b55547a3b1e1e213d3f8b403 : SUCCESS in 2m 08s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/83418146db54428997b1aed7aa4edaa5 : SUCCESS in 7m 24s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a83e256bec6341fb9b2469d7f682489e : SUCCESS in 4m 59s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c63ac414884e4dc483352e9062617ff4 : SUCCESS in 4m 22s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a86829296410435f8601a2ee7eda1c11 : FAILURE in 3m 57s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2900e974cb5043cc9ef323f990663984 : FAILURE in 7m 35s\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b2c90d6819b462c8979ff7c915e1029 : FAILURE in 5m 38s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e90a67ef20d477ab8dc934d820e53ec : FAILURE in 9m 18s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0a8c2c757fd46d7b5a599f00b5a770f : FAILURE in 5m 44s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8dce5bed39b848dab382c058c13423d1 : FAILURE in 4m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9893220b30444eba99f7b802dc5e42a6 : FAILURE in 4m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6c3f269c69443829339d11a9f3310ac : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":2},{"id":"35481e28b088065ca5fc6422dba91a8950c896fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-22 10:42:04.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6a2efdb2693b7a3aa57c263231442be3a8a3196a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-22 11:40:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b802c3d48dae42d6be3ec8e12fa2d431\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae0ee12f72e249859cad2fcebe365be5 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e06271c9c8c648f3944e6ec178f82350 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b672d4311f34eff841a1c80f5b5231f : SUCCESS in 2m 05s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f23893cc8e047a4932b5a8ee00406a2 : SUCCESS in 8m 38s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb62654bb1314240b6038a7944150080 : SUCCESS in 5m 42s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0c55c776c92476cb09c883c95a9367c : SUCCESS in 4m 05s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/070bffdd52e141068e86a8e29e19da9c : SUCCESS in 46m 06s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bec09d8f6a0047a699cbb0b40bf2bf6c : SUCCESS in 44m 11s\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6cbc013d052f4c9a96d48269bf017640 : FAILURE in 24m 42s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/73468ccf1e3b4fe9b3a6ca71730d608e : SUCCESS in 53m 03s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/524bd2f0c3a14a18b6a7577c0f99cd18 : SUCCESS in 13m 45s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80aeff0d0fb943019bec566c3e0a6225 : SUCCESS in 11m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4059326df06488387f8441f49f5e0b8 : SUCCESS in 12m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cf529ed01354b1aa65491658bc24763 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":3},{"id":"cc2548183ff9ce4dee5e50e19b31ca974074b9d3","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-06-22 22:18:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c19f68f578ecf7b55e904263eed0d61e7e1aade0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-06-23 06:48:25.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2c9dd6993a77d151fb712a79b486b6cf545512b8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-06-23 20:44:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ac9690c7e8db6efe870159a1ee69f164dba36370","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 20:44:20.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b13e247f40dc462f9473d68c624639ad\n\n\nWarning:\n  Change 953050,1 in project openstack/puppet-openstack-integration does not share a change queue with 953046 in project openstack/puppet-manila","accounts_in_message":[],"_revision_number":3},{"id":"c2390f309ebecb5968f89d0be5bb86fcdf04b58f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-24 01:08:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6db6ae12df6a630234bcbb9d7d237f94076caff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 01:56:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f095e5fe447424b9a39b4a609af2ab4\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f85ca8bfe29645e6a2e10d7ee14596fe : SUCCESS in 1m 27s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf385186d29c4b42b89b602f978c5817 : SUCCESS in 1m 26s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61c3156a5a4d44319cadbd5db0516bd6 : SUCCESS in 1m 23s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/78c68871de874ea5a23d924a9e2779f4 : SUCCESS in 4m 16s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/768f8793ba0e4812aaae9b9f604d5260 : SUCCESS in 2m 50s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23fa9888317b4f2aa25c5d9cb0b6a9fc : SUCCESS in 2m 44s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44d49a55b56b463f855459d261d9b100 : SUCCESS in 32m 48s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/72d02407cd344cf99eb770ab171d506d : SUCCESS in 43m 18s\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c60550db43e4b778d3532d2548668cd : SUCCESS in 20m 36s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd6938aea00f4457811015d7b96cb04a : SUCCESS in 46m 48s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8ca9eb557b9464893927c26f39b0292 : SUCCESS in 7m 54s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3205b528366f4ee5b2c6f2f70286a605 : SUCCESS in 9m 20s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87cb552d000546f3928aa9be8c2e1611 : SUCCESS in 9m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bc253f6e5f24c91ae0e08b6a35f0dd6 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":3},{"id":"981242a41a1fd0adc3af35964d7677358c25ee22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 01:56:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"582830081f35a77af49dfa3a9f42d903ea71117b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 02:27:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74e0dba10f6745af8a056b6d527b5360\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06789788b2354fb1b1821b913c575344 : SUCCESS in 1m 43s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf2cb98cae654df4846f4692d3720f30 : SUCCESS in 1m 39s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2dd0ffcc11a4322abaa4cd4946a2d7d : SUCCESS in 1m 45s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/77387b83b93b42de9d4423baace165dc : SUCCESS in 4m 26s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6eac3a57ec147dba329efd61c4b8853 : SUCCESS in 2m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5be79a978b745fd9066de8fe20bce1e : SUCCESS in 30m 19s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/587c314006744046b3acf87cebd3a83e : SUCCESS in 28m 05s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/b03977b155a3442cb3cc7bc94495d5bb : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5313a6839b8b4028b915a7e6dfad4574 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":3},{"id":"eaa195dfd34f10e5ed7710205f6f5a669aa37ab0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 02:27:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a476d8c3e6b0c6b7c33be7abc84b6c83b2089761","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 02:28:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e512cfe4c62545069c9efabdaf7b4688\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5a846378120456a9b26214170ccfc03 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"274881b7c47fb07d624f45e8b8db2ed1abf4da7a","revisions":{"7a911b77721b55aab6850159b9418f7b3a9fe429":{"kind":"REWORK","_number":1,"created":"2025-06-22 08:40:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/953046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/46/953046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/46/953046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/46/953046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/46/953046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/46/953046/1"}}},"commit":{"parents":[{"commit":"ecda22e1eefad20ee570e84dd975311d0e9b4e17","subject":"Remove support for backend options in [DEFAULT] section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/ecda22e1eefad20ee570e84dd975311d0e9b4e17"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-22 08:39:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-22 08:39:42.000000000","tz":540},"subject":"Remove management of manila v2 user","message":"Remove management of manila v2 user\n\nIt was deprecated during the previous cycle[1], because manila does not\nrequire the dedicated user for v2 api but the single user is enough.\n\n[1] f2c7a86cfa3f9f460e902a7a6b3e087e1b0f3805\n\nChange-Id: I1063eae7d2ef11da6f1bda2c93e6877e8e8e1d2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/7a911b77721b55aab6850159b9418f7b3a9fe429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/7a911b77721b55aab6850159b9418f7b3a9fe429"}]},"branch":"refs/heads/master"},"3fccbbaf72f7217385cda579b89aa2e184ccea31":{"kind":"REWORK","_number":2,"created":"2025-06-22 08:44:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/953046/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/46/953046/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/46/953046/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/46/953046/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/46/953046/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/46/953046/2"}}},"commit":{"parents":[{"commit":"ecda22e1eefad20ee570e84dd975311d0e9b4e17","subject":"Remove support for backend options in [DEFAULT] section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/ecda22e1eefad20ee570e84dd975311d0e9b4e17"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-22 08:39:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-22 08:44:09.000000000","tz":540},"subject":"Remove management of manila v2 user","message":"Remove management of manila v2 user\n\nIt was deprecated during the previous cycle[1], because manila does not\nrequire the dedicated user for v2 api but the single user is enough.\n\n[1] f2c7a86cfa3f9f460e902a7a6b3e087e1b0f3805\n\nChange-Id: I1063eae7d2ef11da6f1bda2c93e6877e8e8e1d2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/3fccbbaf72f7217385cda579b89aa2e184ccea31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/3fccbbaf72f7217385cda579b89aa2e184ccea31"}]},"branch":"refs/heads/master"},"274881b7c47fb07d624f45e8b8db2ed1abf4da7a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-06-22 10:42:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/953046/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/46/953046/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/46/953046/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/46/953046/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/46/953046/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/46/953046/3"}}},"commit":{"parents":[{"commit":"ecda22e1eefad20ee570e84dd975311d0e9b4e17","subject":"Remove support for backend options in [DEFAULT] section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/ecda22e1eefad20ee570e84dd975311d0e9b4e17"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-22 08:39:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-22 10:41:48.000000000","tz":540},"subject":"Remove management of manila v2 user","message":"Remove management of manila v2 user\n\nIt was deprecated during the previous cycle[1], because manila does not\nrequire the dedicated user for v2 api but the single user is enough.\n\n[1] f2c7a86cfa3f9f460e902a7a6b3e087e1b0f3805\n\nDepends-on: https://review.opendev.org/953050\nChange-Id: I1063eae7d2ef11da6f1bda2c93e6877e8e8e1d2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/274881b7c47fb07d624f45e8b8db2ed1abf4da7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/274881b7c47fb07d624f45e8b8db2ed1abf4da7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
