)]}'
{"id":"openstack%2Fpuppet-manila~958063","triplet_id":"openstack%2Fpuppet-manila~master~Ife51a827fbefb97299b03c5cf5d6826bbd33972e","project":"openstack/puppet-manila","branch":"master","topic":"wsgi-script","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-20 16:15:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ife51a827fbefb97299b03c5cf5d6826bbd33972e","subject":"Replace wsgi script","status":"MERGED","created":"2025-08-20 12:21:06.000000000","updated":"2025-08-20 16:15:39.000000000","submitted":"2025-08-20 16:15:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958063-wsgi-script","meta_rev_id":"c45137fd83077d976b04996ed3ffd5129c772327","_number":958063,"virtual_id_number":958063,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-20 16:15:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-08-20 14:26:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-08-20 15:10:21.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-08-20 15:10:21.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-20 13:30:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-20 14:26:34.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-08-20 15:10:21.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"b2c9d317dd9e13f449638094fba29e48fcb0428f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-20 12:21:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3136b12ca22105ee0b52fbb750ba9eb78e6fbc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 13:30:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8a5776420484499b0d4604c7c0ca26c\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c495d696cee40fca1098637bbe32034 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9697ec8d70434a6c986494d82794e347 : SUCCESS in 2m 55s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a04e5f5e7fc4bcdb2efcc90c3f4f644 : SUCCESS in 6m 42s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3bde2963d7f4762bf454e76a6db2327 : SUCCESS in 4m 11s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f821e810647427a9fbc39872ed2b29a : SUCCESS in 5m 07s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1afd1afcc0547e690d9caf08eee57d3 : SUCCESS in 36m 47s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/acfc5f7da09d472d8ec81907c75f094d : SUCCESS in 58m 59s\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/66d15f755def4221b7049786d983c037 : SUCCESS in 58m 58s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/439a6c3a7f044342b37d624cac50eb9e : FAILURE in 12m 30s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f019762fcf1438890066a489e054c66 : SUCCESS in 1h 02m 46s\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/279428ddf5194015bfcbef6b933bfc5d : SUCCESS in 52m 45s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b8705a65d0144bd82de154686b90736 : SUCCESS in 18m 26s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9c71c51905b4da4b7bdb7760fe81300 : SUCCESS in 9m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8163a5352c154f619adc282f5ce30bb3 : SUCCESS in 13m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d28df26c959da8b11c9be3fed568dccff7a5ab38","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-20 14:26:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5acb90a59b4c33b4d4db092bb5b08ad27c0bb468","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2025-08-20 15:10:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f3f513ae09ab3a60d2a3d8163c2e639e16414639","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 15:11:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8cf8c46e377877b956c80b4d8a2464fa1116a61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 16:15:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bebe2300665f43c8855877844381d555\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18267851c84249a6b409a0d61676351f : SUCCESS in 2m 21s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9809aa2dbc540628574961ae4818851 : SUCCESS in 2m 52s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/202ac3db2a1749de89434aa84f4dd561 : SUCCESS in 5m 16s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61b090fb6bf64b4894b7ef1f7684ae99 : SUCCESS in 6m 50s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eec9803a3d354efb99177a4041ceeddb : SUCCESS in 1h 00m 24s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/793f0a863c0445989178b2f28b59b9ba : SUCCESS in 48m 15s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1dd7ca1f21244629781cee06b903622 : SUCCESS in 19m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c45137fd83077d976b04996ed3ffd5129c772327","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 16:15:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07f361e4b1184cf634ef386003b06e951de17e4f","revisions":{"07f361e4b1184cf634ef386003b06e951de17e4f":{"kind":"REWORK","_number":1,"created":"2025-08-20 12:21:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/958063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/63/958063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/63/958063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/63/958063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/63/958063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/63/958063/1"}}},"commit":{"parents":[{"commit":"990fa781d8ea891bc1fa5f9fb0f9f147967d9895","subject":"Merge \"Remove puppet 7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/990fa781d8ea891bc1fa5f9fb0f9f147967d9895"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-20 12:20:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-20 12:20:56.000000000","tz":540},"subject":"Replace wsgi script","message":"Replace wsgi script\n\npbr is removing support for wsgi_script due to changes in setuptools.\nUse the wsgi script module python code instead.\n\nNote that this change only touches CentOS code. Ubuntu and Debian will\nbe updated later when we bump the release used in our testing.\n\nChange-Id: Ife51a827fbefb97299b03c5cf5d6826bbd33972e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/07f361e4b1184cf634ef386003b06e951de17e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/07f361e4b1184cf634ef386003b06e951de17e4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
