)]}'
{"id":"openstack%2Fpuppet-manila~962745","triplet_id":"openstack%2Fpuppet-manila~master~I89bf0fe28b354d76386d09a96fa1e47998d1596c","project":"openstack/puppet-manila","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-03 14:49:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I89bf0fe28b354d76386d09a96fa1e47998d1596c","subject":"Allow customizing format of service image","status":"MERGED","created":"2025-10-02 06:51:24.000000000","updated":"2025-10-03 14:50:27.000000000","submitted":"2025-10-03 14:49:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962745","meta_rev_id":"eb3fcfef26cdd1d9bcdc67374bb6655ee5e1d748","_number":962745,"virtual_id_number":962745,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-03 14:49:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-10-02 21:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-10-03 13:43:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-10-03 13:43:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-02 09:04:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-02 21:05:01.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-10-03 13:43:41.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"f9d5d568dc4e19cb43eb127f52f06ba319c0117b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 06:51:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ade36dcf36e7da23032e843ac62c64eb0f4d26b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 07:13:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6bea39e8c229cea4797f609ae8eb8c7eb3c0fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 07:18:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9688eacf18ed3a935c6ccbbc636ff997d5b61deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 07:36:31.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"81ede912c1ebd43e69394c95fc4c1b8fe3a05845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 07:59:23.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"cb21ed755a88ba091345c94fddcdd972929061c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 09:04:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51ba5984ce484e5da4addc5fd21b5180\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f918ce0684ad41b7ae8dff273a3251af : SUCCESS in 2m 48s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcab4cf8b9704759a6bd9127c17caa95 : SUCCESS in 2m 50s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/b7287afd1d1749e8809f2ffdb2bfa1ed : FAILURE in 6m 00s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecfa44cc4a7e488d9f626ed7c10965ce : FAILURE in 4m 33s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/623e2e7bf55d47199ecd030bca5a3c43 : FAILURE in 3m 13s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/672913e5ab574069a52392c2faf93a75 : SUCCESS in 39m 35s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5402f782b3f4ccc9289b98967ff20e5 : SUCCESS in 53m 38s\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/57c5cd1b873a40bf8c2724c2e3358131 : SUCCESS in 32m 21s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec9946e0690c482492ec1144671754f2 : FAILURE in 43m 56s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/695e419e64bf452490cab32ff0f61247 : SUCCESS in 1h 00m 03s\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2de0f158b02b42fb9a7ea51ba3071d47 : SUCCESS in 54m 23s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/323746afeec540a4b3e15cbe1d0c426c : SUCCESS in 17m 03s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89eea842fccd45528fc140b5144d6d8c : SUCCESS in 12m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/830706bbe5ad424ba549bde4c23f2e02 : SUCCESS in 6m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffbae7814004426f9d72ab2fd4377efc : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":5},{"id":"debec420a06c3341fb30dd4d39acdd5380c43594","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 09:10:28.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0e352bf3a4b383f5de4a6a70f28cd65177fcdcc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 10:15:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4313f820f50f4271aa71cbb0d1b7d302\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f238b4f5c8fd42cea1f525ac5c0902a2 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72e456ac646e43eeaafd3a4a5a6fe0e8 : SUCCESS in 2m 38s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/38404f2a13244e82b89f745e10fee36c : FAILURE in 5m 24s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c476762b837e4ca5a0c3e0014350e3f9 : FAILURE in 6m 01s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bf0e2c1acd241d99e4eba5dfe68ae03 : FAILURE in 5m 55s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2c39c108da74b05846e6266156abea3 : SUCCESS in 41m 34s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7b2a99210ad4109abc4f1a21eef53d8 : SUCCESS in 57m 58s\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b7395fec13642ea879ae7a8d076420f : SUCCESS in 57m 42s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74b2407758194056a849486c34615486 : FAILURE in 49m 51s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5940efde07384e9387d52bef2571b57c : SUCCESS in 58m 41s\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d911e74679734d83885c85adad81c69d : SUCCESS in 30m 54s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/496c2e09fb3f42f89d11b7c09330d743 : SUCCESS in 17m 59s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b7b7ae7098d447caa599d6f35df7113 : SUCCESS in 12m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b950a64363da4b8db60df9c36794d1f9 : SUCCESS in 12m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54a612e206924643ad9f8e2f0f75972e : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":6},{"id":"e44a9c4e366f4cd65df462139038ef34c679ed6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 10:48:27.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"803671fc1e1408026601c456a236b0891f1799b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 11:48:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a7e611d7bc94e9a96eb7b4b63eb4d73\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20011c695ee04b8dbe7e1c626fa1b8fc : SUCCESS in 2m 51s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/133717c3cc4945469ac776d151f3522a : SUCCESS in 2m 45s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/176ececc7d694103b1c61bb22d31f63f : FAILURE in 5m 00s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1f8458d5c0241988cf1ec4e30715e99 : FAILURE in 5m 47s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52240bfd6ca843308d6f37f5c9ed1f58 : FAILURE in 3m 45s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ccc44d9f6cb4929918cb4eaaed7aea1 : SUCCESS in 41m 02s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fd2afdcae524ab591440c247805bc01 : SUCCESS in 50m 06s\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/71a04ebe94054412bd7f42c9c246ed7e : SUCCESS in 55m 01s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d285ba077bc4e278ee423ad601fecad : FAILURE in 48m 16s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/72147374cf8e4f59b71dfe32850972e5 : SUCCESS in 55m 04s\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6bca5a2de4134b62abcdcae0240e0eeb : SUCCESS in 33m 52s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d20463d952f4cfcb17b39a9b365e07c : SUCCESS in 13m 03s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99ab643d8a384bf2b0d7798edf773256 : SUCCESS in 8m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73b9132f3eae49c4b6f8c4a9b69d3c40 : SUCCESS in 7m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db0d0ac4f93146abad42c3f11be7aa9e : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":7},{"id":"d279a9770b2336528c88fd7d73645868123420d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 11:50:11.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"01737682f6fee94a2edbd5b13d1be16b4ab748bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 12:51:21.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c27d2869c0645c1a6f153acb1223a06\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e87be8d6f56d4a258b706475ae4f7493 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a08ea82191b4cfeadef316b195c05f8 : SUCCESS in 1m 48s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/d177cc7f26c74601b963ab40b532da4e : SUCCESS in 6m 25s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4c8b9da60104811ac135dfdaec29922 : SUCCESS in 5m 55s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c73c49ebaac4f349f0ea2d1fd9cb1af : SUCCESS in 5m 57s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/637ed0bb96844fe7a4a8d0b9a1d2f81d : SUCCESS in 46m 52s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb70ea3c6c9b4c90bbbf2f159900d76c : SUCCESS in 31m 40s\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b95899ad71d34e94a2deb6f0c4c20d88 : SUCCESS in 33m 15s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b64d559e5e304f39b0c77694a115cdf9 : FAILURE in 34m 39s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/789ebede95794bfdab7772fab0bd25f1 : SUCCESS in 56m 18s\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35ebc92520344a0ebd101e37b3dbba2a : SUCCESS in 56m 23s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3fcc012d3b8c4c8b8e32cde8fc268dde : SUCCESS in 17m 42s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0792b6a189164a43ab8050f56554d94c : SUCCESS in 9m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6dee0eaf65ee4852b92389f4beab5215 : SUCCESS in 7m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/209849bd0d3e43629e85eaac5f264484 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":8},{"id":"d75e6ba811c1168a2635872e00171b0668d0140d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-02 21:05:01.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"34fd4cea439edba622c474951cb223d2ed0a8f05","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-10-03 13:43:41.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"f0ac49c05aae18c6640584f41ae8319d7b91346e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 13:44:29.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"d04b9560d3f72ea438396b0285929a83be974be9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 14:49:34.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdb199d622514e56bfda5aee515649ec\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f07a3ca249344aa3a02d86fc0ab537c9 : SUCCESS in 5m 03s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1f567b402194ad09c5f49f9cfddac27 : SUCCESS in 2m 54s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/843cdef846864140a6592c8abf1cb917 : SUCCESS in 4m 27s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dac4dc6608104b3fb724b4b0ec08a0eb : SUCCESS in 6m 17s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0136954e50343ccb57cc4e41bc954b4 : SUCCESS in 55m 40s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ff7f386a9144c2494fa2b39a8396e1a : SUCCESS in 59m 51s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdeb6f248751425c9bdfc3ed78557e97 : SUCCESS in 17m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41f59f95fd4b4f959ef9db57ccf9312c : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":8},{"id":"d59dada78ebf985c2e3e5f3da193b9ce81acb580","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 14:49:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"eb3fcfef26cdd1d9bcdc67374bb6655ee5e1d748","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 14:50:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c8c79ebd9414915b1a9416bf8ae9e36\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/188d08e73924453aad26258b1bc2ca29 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"8a8543e92b9cf7f412b23c029729224557781137","revisions":{"b245bd9867f43d8e867317140d0777d249e89e1f":{"kind":"REWORK","_number":1,"created":"2025-10-02 06:51:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/962745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/45/962745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/1"}}},"commit":{"parents":[{"commit":"394432804991b3a806b33577f71cf870fdcc856d","subject":"Fix incomplete assertions of manila::backend::service_instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/394432804991b3a806b33577f71cf870fdcc856d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 06:50:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 06:51:08.000000000","tz":540},"subject":"Allow customizing format of service image","message":"Allow customizing format of service image\n\nThis is useful especially when raw disk format is preferred to\nleverage in-storage copy functionality.\n\nChange-Id: I89bf0fe28b354d76386d09a96fa1e47998d1596c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/b245bd9867f43d8e867317140d0777d249e89e1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/b245bd9867f43d8e867317140d0777d249e89e1f"}]},"branch":"refs/heads/master"},"269fd6e90a2be96275142662350c88c09b87ffb9":{"kind":"REWORK","_number":2,"created":"2025-10-02 07:13:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/962745/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/45/962745/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/2"}}},"commit":{"parents":[{"commit":"e182ca9e8749a8de6bd3f7e7f8fc62833285ab43","subject":"Fix incomplete assertions of manila::backend::service_instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/e182ca9e8749a8de6bd3f7e7f8fc62833285ab43"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 06:50:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 07:13:21.000000000","tz":540},"subject":"Allow customizing format of service image","message":"Allow customizing format of service image\n\nThis is useful especially when raw disk format is preferred to\nleverage in-storage copy functionality.\n\nChange-Id: I89bf0fe28b354d76386d09a96fa1e47998d1596c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/269fd6e90a2be96275142662350c88c09b87ffb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/269fd6e90a2be96275142662350c88c09b87ffb9"}]},"branch":"refs/heads/master"},"e55f3bcf6b750f94ddb2d08b4c37efba2fae26b5":{"kind":"REWORK","_number":3,"created":"2025-10-02 07:18:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/962745/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/45/962745/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/3"}}},"commit":{"parents":[{"commit":"b6b03884c9b6157cc82d00be3facdb3ccd7606d5","subject":"Fix incomplete assertions of manila::backend::service_instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/b6b03884c9b6157cc82d00be3facdb3ccd7606d5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 06:50:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 07:18:16.000000000","tz":540},"subject":"Allow customizing format of service image","message":"Allow customizing format of service image\n\nThis is useful especially when raw disk format is preferred to\nleverage in-storage copy functionality.\n\nChange-Id: I89bf0fe28b354d76386d09a96fa1e47998d1596c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/e55f3bcf6b750f94ddb2d08b4c37efba2fae26b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/e55f3bcf6b750f94ddb2d08b4c37efba2fae26b5"}]},"branch":"refs/heads/master"},"6b0734c64bce979d35253c9df4c258c5a4302319":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-10-02 07:36:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/962745/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/45/962745/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/4"}}},"commit":{"parents":[{"commit":"231855c6c1f44f2f7b5104c12fc4aee374715110","subject":"Fix incomplete assertions of manila::backend::service_instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/231855c6c1f44f2f7b5104c12fc4aee374715110"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 06:50:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 07:36:24.000000000","tz":540},"subject":"Allow customizing format of service image","message":"Allow customizing format of service image\n\nThis is useful especially when raw disk format is preferred to\nleverage in-storage copy functionality.\n\nChange-Id: I89bf0fe28b354d76386d09a96fa1e47998d1596c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/6b0734c64bce979d35253c9df4c258c5a4302319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/6b0734c64bce979d35253c9df4c258c5a4302319"}]},"branch":"refs/heads/master"},"daf956a549d87c6c64293184c45b054b46817db1":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-10-02 07:59:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/962745/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/45/962745/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/5"}}},"commit":{"parents":[{"commit":"3ed8b8bbe5e63eb142f9ed1e32c504afdf612d03","subject":"Fix incomplete assertions of manila::backend::service_instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/3ed8b8bbe5e63eb142f9ed1e32c504afdf612d03"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 06:50:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 07:59:11.000000000","tz":540},"subject":"Allow customizing format of service image","message":"Allow customizing format of service image\n\nThis is useful especially when raw disk format is preferred to\nleverage in-storage copy functionality.\n\nChange-Id: I89bf0fe28b354d76386d09a96fa1e47998d1596c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/daf956a549d87c6c64293184c45b054b46817db1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/daf956a549d87c6c64293184c45b054b46817db1"}]},"branch":"refs/heads/master"},"9f4b64afb11e0f6b4c7da11378c092f94f592c8a":{"kind":"REWORK","_number":6,"created":"2025-10-02 09:10:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/962745/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/45/962745/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/6"}}},"commit":{"parents":[{"commit":"f625bc858cce7d7ad3038dc9e1c00f75cd492ce0","subject":"Fix incomplete assertions of manila::backend::service_instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f625bc858cce7d7ad3038dc9e1c00f75cd492ce0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 06:50:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 09:10:17.000000000","tz":540},"subject":"Allow customizing format of service image","message":"Allow customizing format of service image\n\nThis is useful especially when raw disk format is preferred to\nleverage in-storage copy functionality.\n\nChange-Id: I89bf0fe28b354d76386d09a96fa1e47998d1596c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9f4b64afb11e0f6b4c7da11378c092f94f592c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/9f4b64afb11e0f6b4c7da11378c092f94f592c8a"}]},"branch":"refs/heads/master"},"b0fc12f830703af6d5e9388ccb91c493ece4a2e7":{"kind":"REWORK","_number":7,"created":"2025-10-02 10:48:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/962745/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/45/962745/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/7"}}},"commit":{"parents":[{"commit":"f625bc858cce7d7ad3038dc9e1c00f75cd492ce0","subject":"Fix incomplete assertions of manila::backend::service_instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f625bc858cce7d7ad3038dc9e1c00f75cd492ce0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 06:50:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 10:48:27.000000000","tz":0},"subject":"Allow customizing format of service image","message":"Allow customizing format of service image\n\nThis is useful especially when raw disk format is preferred to\nleverage in-storage copy functionality.\n\nChange-Id: I89bf0fe28b354d76386d09a96fa1e47998d1596c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/b0fc12f830703af6d5e9388ccb91c493ece4a2e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/b0fc12f830703af6d5e9388ccb91c493ece4a2e7"}]},"branch":"refs/heads/master"},"8a8543e92b9cf7f412b23c029729224557781137":{"kind":"REWORK","_number":8,"created":"2025-10-02 11:50:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/962745/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-manila","ref":"refs/changes/45/962745/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-manila refs/changes/45/962745/8"}}},"commit":{"parents":[{"commit":"f625bc858cce7d7ad3038dc9e1c00f75cd492ce0","subject":"Fix incomplete assertions of manila::backend::service_instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/f625bc858cce7d7ad3038dc9e1c00f75cd492ce0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 06:50:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 11:50:02.000000000","tz":540},"subject":"Allow customizing format of service image","message":"Allow customizing format of service image\n\nThis is useful especially when raw disk format is preferred to\nleverage in-storage copy functionality.\n\nChange-Id: I89bf0fe28b354d76386d09a96fa1e47998d1596c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/8a8543e92b9cf7f412b23c029729224557781137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-manila/commit/8a8543e92b9cf7f412b23c029729224557781137"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
