)]}'
{"id":"openstack%2Fpuppet-mistral~659408","triplet_id":"openstack%2Fpuppet-mistral~master~I2d15a0b30394e9187bda81f5dcb0074d579574c5","project":"openstack/puppet-mistral","branch":"master","topic":"add_openstackclient","hashtags":[],"change_id":"I2d15a0b30394e9187bda81f5dcb0074d579574c5","subject":"Add openstackclient installation to the client class","status":"MERGED","created":"2019-05-16 01:53:06.000000000","updated":"2019-05-20 02:38:24.000000000","submitted":"2019-05-20 02:38:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659408-1558319904827-978e37b0","meta_rev_id":"4dd4cf64bf24021d109e5510546e135e39d2c5e3","_number":659408,"virtual_id_number":659408,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-20 02:38:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-17 09:56:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-05-17 19:27:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-05-20 01:04:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 09:56:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-17 19:27:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-20 02:38:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"223de907360def7a80eeed0a174db3e93f149d3a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-16 01:53:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9775f5515fca25bba35b3210e295bd67749a04d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 03:26:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/08/659408/1/check/puppet-openstack-lint-ubuntu-bionic/a50d7e1/ : SUCCESS in 3m 51s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/08/659408/1/check/puppet-openstack-syntax-4-ubuntu-bionic/9c27140/ : SUCCESS in 3m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/08/659408/1/check/puppet-openstack-syntax-5-ubuntu-bionic/3aeeaaf/ : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/08/659408/1/check/puppet-openstack-syntax-6-ubuntu-bionic/0f565b8/ : SUCCESS in 7m 03s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/08/659408/1/check/puppet-openstack-unit-4.8-centos-7/1c2bdf4/ : SUCCESS in 20m 46s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/08/659408/1/check/puppet-openstack-unit-5.5-centos-7/7e52804/ : SUCCESS in 22m 44s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/08/659408/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/44dd0d3/ : SUCCESS in 9m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/08/659408/1/check/puppet-openstack-unit-latest-ubuntu-bionic/ae0b27f/ : SUCCESS in 6m 50s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/08/659408/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/6e553e0/ : FAILURE in 23m 44s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/08/659408/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/45e1610/ : SUCCESS in 1h 04m 41s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/08/659408/1/check/puppet-openstack-beaker-centos-7/6378aa2/ : SUCCESS in 18m 15s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/08/659408/1/check/puppet-openstack-beaker-ubuntu-bionic/5cb4488/ : FAILURE in 7m 40s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/08/659408/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/fda140a/ : SUCCESS in 1h 21m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"988337dd39f924b5635c14acc8d534f929a6993c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-16 03:35:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/08/659408/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/74ce05e/ : SUCCESS in 1h 34m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/08/659408/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/4d3a165/ : SUCCESS in 1h 39m 37s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/08/659408/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/0b2038a/ : SUCCESS in 1h 38m 02s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/08/659408/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/53e1cf0/ : FAILURE in 1h 15m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e2dab02a8b42e0f0a4741c46480f5275697c2be","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-17 01:01:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"93e33d3650935b4cd9eb5ce3da787229423898fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 02:33:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/08/659408/2/check/puppet-openstack-lint-ubuntu-bionic/72a880c/ : SUCCESS in 4m 32s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/08/659408/2/check/puppet-openstack-syntax-4-ubuntu-bionic/ac40b12/ : SUCCESS in 4m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/08/659408/2/check/puppet-openstack-syntax-5-ubuntu-bionic/7c4eb6d/ : SUCCESS in 5m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/08/659408/2/check/puppet-openstack-syntax-6-ubuntu-bionic/be8adab/ : SUCCESS in 3m 48s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/08/659408/2/check/puppet-openstack-unit-4.8-centos-7/0eaf585/ : SUCCESS in 16m 56s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/08/659408/2/check/puppet-openstack-unit-5.5-centos-7/97aed2b/ : SUCCESS in 12m 04s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/08/659408/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/6d03dc8/ : SUCCESS in 6m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/08/659408/2/check/puppet-openstack-unit-latest-ubuntu-bionic/1bee89a/ : SUCCESS in 8m 07s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/08/659408/2/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/c667657/ : FAILURE in 22m 44s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/08/659408/2/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/efce63e/ : SUCCESS in 1h 00m 22s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/08/659408/2/check/puppet-openstack-beaker-centos-7/0be1c8e/ : SUCCESS in 15m 41s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/08/659408/2/check/puppet-openstack-beaker-ubuntu-bionic/1dcc706/ : FAILURE in 6m 54s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/08/659408/2/check/tripleo-puppet-ci-centos-7-undercloud-containers/697ac58/ : SUCCESS in 1h 25m 42s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/08/659408/2/check/build-openstack-releasenotes/e3c3b61/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c33a8770f47a4a98bad15b35686b42a4fe063a37","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-17 02:39:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/08/659408/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/dcb6947/ : SUCCESS in 1h 31m 05s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/08/659408/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/ccf3a65/ : SUCCESS in 1h 34m 30s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/08/659408/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/d24c2e8/ : SUCCESS in 1h 35m 34s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/08/659408/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/dc03706/ : SUCCESS in 1h 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a53d3d120b4826b89d6edcd215c87ee1fea453c7","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-17 08:32:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"76f687cfb77a26ecaa6eba05e60934524c0fd8cc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-17 09:56:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/08/659408/3/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/367c043/ : SUCCESS in 1h 19m 16s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/08/659408/3/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/3068018/ : FAILURE in 1h 21m 20s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/08/659408/3/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/a2070a1/ : SUCCESS in 1h 19m 10s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/08/659408/3/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/27b7e73/ : SUCCESS in 54m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4aded7b5fe10246ac422084709257a6220d65891","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 10:09:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/08/659408/3/check/puppet-openstack-lint-ubuntu-bionic/666589e/ : SUCCESS in 4m 46s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/08/659408/3/check/puppet-openstack-syntax-4-ubuntu-bionic/a4ad112/ : SUCCESS in 4m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/08/659408/3/check/puppet-openstack-syntax-5-ubuntu-bionic/86321df/ : SUCCESS in 5m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/08/659408/3/check/puppet-openstack-syntax-6-ubuntu-bionic/de6c7f2/ : SUCCESS in 6m 31s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/08/659408/3/check/puppet-openstack-unit-4.8-centos-7/cb92c9c/ : SUCCESS in 17m 50s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/08/659408/3/check/puppet-openstack-unit-5.5-centos-7/990f2ec/ : SUCCESS in 16m 59s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/08/659408/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/e5b4219/ : SUCCESS in 10m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/08/659408/3/check/puppet-openstack-unit-latest-ubuntu-bionic/bde825e/ : SUCCESS in 7m 34s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/08/659408/3/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/e5156db/ : FAILURE in 30m 22s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/08/659408/3/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/a0b7fab/ : SUCCESS in 58m 23s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/08/659408/3/check/puppet-openstack-beaker-centos-7/e4af471/ : SUCCESS in 16m 21s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/08/659408/3/check/puppet-openstack-beaker-ubuntu-bionic/85b85dc/ : FAILURE in 6m 52s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/08/659408/3/check/tripleo-puppet-ci-centos-7-undercloud-containers/30bb547/ : SUCCESS in 1h 32m 19s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/08/659408/3/check/build-openstack-releasenotes/2b267a9/html/ : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":3},{"id":"789a46a6a94dc14a34cd76600407bed9c231a1f7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-17 19:27:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c8f5b3401dc45addd76b464b5bde8ed0e1a6e6e1","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 01:04:59.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8bcbee03aec0738c067c81530c7a069254619827","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 01:05:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d0abb0b4e64621a67c11829218bbdfa7592473d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 02:38:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/08/659408/3/gate/puppet-openstack-lint-ubuntu-bionic/b937722/ : SUCCESS in 4m 44s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/08/659408/3/gate/puppet-openstack-syntax-4-ubuntu-bionic/bc6b82d/ : SUCCESS in 6m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/08/659408/3/gate/puppet-openstack-syntax-5-ubuntu-bionic/ddd48ed/ : SUCCESS in 6m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/08/659408/3/gate/puppet-openstack-syntax-6-ubuntu-bionic/dc21b5d/ : SUCCESS in 4m 26s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/08/659408/3/gate/puppet-openstack-unit-4.8-centos-7/5771d9e/ : SUCCESS in 13m 26s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/08/659408/3/gate/puppet-openstack-unit-5.5-centos-7/6cf6e58/ : SUCCESS in 12m 31s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/08/659408/3/gate/puppet-openstack-unit-6.4-ubuntu-bionic/08ddca8/ : SUCCESS in 8m 50s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/08/659408/3/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/139d090/ : SUCCESS in 54m 39s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/08/659408/3/gate/puppet-openstack-beaker-centos-7/bd8d99d/ : SUCCESS in 15m 45s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/08/659408/3/gate/puppet-openstack-beaker-ubuntu-bionic/7bf9e6e/ : FAILURE in 8m 33s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/08/659408/3/gate/tripleo-puppet-ci-centos-7-undercloud-containers/aa2fc18/ : SUCCESS in 1h 28m 56s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/08/659408/3/gate/build-openstack-releasenotes/3d10fbc/html/ : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":3},{"id":"da60b872758802ca4bee8d04a4636605dbeb1ffd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 02:38:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b33ccfd5365e145b3c870e5264fac8e8d27cfdcc","revisions":{"fdd78060720fea14d5a2be1024bdd27944f78535":{"kind":"REWORK","_number":1,"created":"2019-05-16 01:53:06.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/08/659408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/08/659408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/08/659408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/08/659408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/08/659408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/08/659408/1"}}},"commit":{"parents":[{"commit":"7b2a9b0a4e47fc943c35a18d44917c44888ba808","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/7b2a9b0a4e47fc943c35a18d44917c44888ba808"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:24.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:24.000000000","tz":480},"subject":"Add openstackclient","message":"Add openstackclient\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I2d15a0b30394e9187bda81f5dcb0074d579574c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/fdd78060720fea14d5a2be1024bdd27944f78535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/fdd78060720fea14d5a2be1024bdd27944f78535"}]},"branch":"refs/heads/master"},"458791fd8bf171d1878bfd977796e59b75fb0beb":{"kind":"REWORK","_number":2,"created":"2019-05-17 01:01:43.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/08/659408/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/08/659408/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/08/659408/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/08/659408/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/08/659408/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/08/659408/2"}}},"commit":{"parents":[{"commit":"7b2a9b0a4e47fc943c35a18d44917c44888ba808","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/7b2a9b0a4e47fc943c35a18d44917c44888ba808"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:24.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-17 01:00:21.000000000","tz":480},"subject":"Add openstackclient installation to the client class","message":"Add openstackclient installation to the client class\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I2d15a0b30394e9187bda81f5dcb0074d579574c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/458791fd8bf171d1878bfd977796e59b75fb0beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/458791fd8bf171d1878bfd977796e59b75fb0beb"}]},"branch":"refs/heads/master"},"b33ccfd5365e145b3c870e5264fac8e8d27cfdcc":{"kind":"REWORK","_number":3,"created":"2019-05-17 08:32:34.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/08/659408/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/08/659408/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/08/659408/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/08/659408/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/08/659408/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/08/659408/3"}}},"commit":{"parents":[{"commit":"7b2a9b0a4e47fc943c35a18d44917c44888ba808","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/7b2a9b0a4e47fc943c35a18d44917c44888ba808"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:24.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-17 08:31:18.000000000","tz":480},"subject":"Add openstackclient installation to the client class","message":"Add openstackclient installation to the client class\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I2d15a0b30394e9187bda81f5dcb0074d579574c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b33ccfd5365e145b3c870e5264fac8e8d27cfdcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b33ccfd5365e145b3c870e5264fac8e8d27cfdcc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
