)]}'
{"id":"openstack%2Fpuppet-mistral~663198","triplet_id":"openstack%2Fpuppet-mistral~master~I0802adc0d5d3811f08a52df7e5b4f4a19366813f","project":"openstack/puppet-mistral","branch":"master","topic":"prep-release-train-1","hashtags":[],"change_id":"I0802adc0d5d3811f08a52df7e5b4f4a19366813f","subject":"Prepare Train M2","status":"MERGED","created":"2019-06-05 01:42:56.000000000","updated":"2019-06-10 13:30:48.000000000","submitted":"2019-06-10 13:30:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663198-1560173448049-811cd442","meta_rev_id":"3c73b5aaa90755a60aeaf2a1b3f2b4a1f93c09d6","_number":663198,"virtual_id_number":663198,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-10 13:30:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-06-10 13:20:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-06-10 13:20:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 13:20:41.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-10 13:30:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"50364de175a6a5638d62d88fc0f7f33f9cf79e0d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-05 01:42:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6936bd7cae12378cc9fa9b4902ba3aa9b19371ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 02:03:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/98/663198/1/check/build-openstack-puppet-tarball/999bea3/ : SUCCESS in 1m 51s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/98/663198/1/check/puppet-openstack-lint-ubuntu-bionic/7f23dae/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"eeb7db7c045018456b07cbaeb97649f25d2fbb8d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-10 02:09:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4740c6968f27a2ef4d1859c63c9903a98a7cc7c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 02:14:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/98/663198/2/check/build-openstack-puppet-tarball/44c830a/ : SUCCESS in 1m 43s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/98/663198/2/check/puppet-openstack-lint-ubuntu-bionic/eea1e7e/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"dac7fdc0804c9b71c194c2fed2a5e6f17f0a6714","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 13:20:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"11fb8b9c952b7643d8ed65f1760ad1da8fa27b83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:20:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8a3a64bef201f8e8d4594ac1955810d125f337d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:30:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/98/663198/2/gate/build-openstack-puppet-tarball/6adc029/ : SUCCESS in 2m 04s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/98/663198/2/gate/puppet-openstack-lint-ubuntu-bionic/29d7472/ : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"3b32f145af6e7ca58f4c14485ef8b5a570441e6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:30:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"030a4fe27e2c32e6559dd89d25ce0132f1c252d6","revisions":{"92c8fa46b72ece19c5b82644540c9aef0b5a3d50":{"kind":"REWORK","_number":1,"created":"2019-06-05 01:42:56.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/98/663198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/98/663198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/98/663198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/98/663198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/98/663198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/98/663198/1"}}},"commit":{"parents":[{"commit":"b33ccfd5365e145b3c870e5264fac8e8d27cfdcc","subject":"Add openstackclient installation to the client class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b33ccfd5365e145b3c870e5264fac8e8d27cfdcc"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:20.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:20.000000000","tz":480},"subject":"Prepare Train M1","message":"Prepare Train M1\n\nUpdate the version metadata for Train milestone 1\n\nChange-Id: I0802adc0d5d3811f08a52df7e5b4f4a19366813f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/92c8fa46b72ece19c5b82644540c9aef0b5a3d50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/92c8fa46b72ece19c5b82644540c9aef0b5a3d50"}]},"branch":"refs/heads/master"},"030a4fe27e2c32e6559dd89d25ce0132f1c252d6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-10 02:09:21.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/98/663198/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/98/663198/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/98/663198/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/98/663198/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/98/663198/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/98/663198/2"}}},"commit":{"parents":[{"commit":"b33ccfd5365e145b3c870e5264fac8e8d27cfdcc","subject":"Add openstackclient installation to the client class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/b33ccfd5365e145b3c870e5264fac8e8d27cfdcc"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:20.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-10 02:04:02.000000000","tz":480},"subject":"Prepare Train M2","message":"Prepare Train M2\n\nUpdate the version metadata for Train milestone 2\n\nChange-Id: I0802adc0d5d3811f08a52df7e5b4f4a19366813f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/030a4fe27e2c32e6559dd89d25ce0132f1c252d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/030a4fe27e2c32e6559dd89d25ce0132f1c252d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
