)]}'
{"id":"openstack%2Fpuppet-mistral~669034","triplet_id":"openstack%2Fpuppet-mistral~master~I3a658d4426689ba09188b0889082b63a744065fc","project":"openstack/puppet-mistral","branch":"master","topic":"bug/1835161","hashtags":[],"change_id":"I3a658d4426689ba09188b0889082b63a744065fc","subject":"Manage oslo_middleware/max_request_body_size","status":"MERGED","created":"2019-07-04 03:44:21.000000000","updated":"2019-07-05 02:34:39.000000000","submitted":"2019-07-05 02:34:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669034-1562294079656-5870c8eb","meta_rev_id":"1fec42e57cab8c49710e32f1dfd622f0f9163969","_number":669034,"virtual_id_number":669034,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-07-05 02:34:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-04 05:15:05.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-04 13:59:53.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-05 00:51:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 05:15:05.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-04 13:59:53.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-05 02:34:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd8808e2676cf3ce6a19f0517cafa564249f36c2","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-04 03:44:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06e0ea934dfd5d2c9bdc971db4f598884c96fab2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-04 05:15:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/34/669034/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/6cbdbaa/ : SUCCESS in 1h 21m 30s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/34/669034/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/8e34ace/ : SUCCESS in 1h 27m 07s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/34/669034/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/49a1b70/ : FAILURE in 55m 17s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/34/669034/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/5c29f91/ : SUCCESS in 40m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb299d1a49db35075ee69fce16fd54bb30590771","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 05:31:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/34/669034/1/check/puppet-openstack-lint-ubuntu-bionic/51fc47d/ : SUCCESS in 5m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/34/669034/1/check/puppet-openstack-syntax-5-ubuntu-bionic/c84fae6/ : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/34/669034/1/check/puppet-openstack-syntax-6-ubuntu-bionic/fa6b8f5/ : SUCCESS in 4m 09s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/34/669034/1/check/puppet-openstack-unit-5.5-centos-7/672689c/ : SUCCESS in 11m 22s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/34/669034/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/359073e/ : SUCCESS in 11m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/34/669034/1/check/puppet-openstack-unit-latest-ubuntu-bionic/cf830ef/ : SUCCESS in 13m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/34/669034/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/623d775/ : FAILURE in 29m 39s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/34/669034/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/a987e68/ : SUCCESS in 58m 41s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/34/669034/1/check/puppet-openstack-beaker-centos-7/8b42050/ : SUCCESS in 19m 27s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/34/669034/1/check/puppet-openstack-beaker-ubuntu-bionic/f925595/ : FAILURE in 8m 45s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/34/669034/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/5b0815b/ : SUCCESS in 1h 38m 58s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/34/669034/1/check/build-openstack-releasenotes/6d921b7/html/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"fa1def3b7cd1bc3c36575fec04bc3cf68f64eb21","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-04 13:59:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b31605c373956ea6365080cf0229b13b2e49d261","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-05 00:51:27.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d1fe76c7ba44748221737e46ba7a987da50fb5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 00:51:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5aa4ead54c010e49d5eeaf6c6f421521ca71dd1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 02:34:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7eaf4b4ba6d452efca27a37d12588fce8d2e7686","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 02:34:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/34/669034/1/gate/puppet-openstack-lint-ubuntu-bionic/5a97fe8/ : SUCCESS in 4m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/34/669034/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/0bb0772/ : SUCCESS in 5m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/34/669034/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/122cb3d/ : SUCCESS in 4m 01s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/34/669034/1/gate/puppet-openstack-unit-5.5-centos-7/da37f03/ : SUCCESS in 11m 15s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/34/669034/1/gate/puppet-openstack-unit-6.4-ubuntu-bionic/35ac26c/ : SUCCESS in 8m 56s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/34/669034/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/ba2f91e/ : SUCCESS in 1h 02m 39s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/34/669034/1/gate/puppet-openstack-beaker-centos-7/55f6d00/ : SUCCESS in 15m 44s\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/34/669034/1/gate/tripleo-puppet-ci-centos-7-undercloud-containers/db9e91e/ : SUCCESS in 1h 42m 39s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/34/669034/1/gate/build-openstack-releasenotes/a76eb00/html/ : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a28a993c98cbeb3328eadbb3c0b63e3e60e2e25","revisions":{"8a28a993c98cbeb3328eadbb3c0b63e3e60e2e25":{"kind":"REWORK","_number":1,"created":"2019-07-04 03:44:21.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/34/669034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/34/669034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/34/669034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/34/669034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/34/669034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/34/669034/1"}}},"commit":{"parents":[{"commit":"030a4fe27e2c32e6559dd89d25ce0132f1c252d6","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/030a4fe27e2c32e6559dd89d25ce0132f1c252d6"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-04 03:34:43.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-04 03:36:57.000000000","tz":480},"subject":"Manage oslo_middleware/max_request_body_size","message":"Manage oslo_middleware/max_request_body_size\n\nSo that we can increase it from the default 114688\n\nUseful in case for example the OS-Federation mapping is too large.\n\nIf this limit is breached mistral will return a 413 Entity Too Large\nand not log anything to mistral.log.\n\nChange-Id: I3a658d4426689ba09188b0889082b63a744065fc\nCloses-Bug: #1835161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/8a28a993c98cbeb3328eadbb3c0b63e3e60e2e25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/8a28a993c98cbeb3328eadbb3c0b63e3e60e2e25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
