)]}'
{"id":"openstack%2Fpuppet-mistral~684903","triplet_id":"openstack%2Fpuppet-mistral~master~Id4fa0518ab7fb7d60f0d142666709d233db1966e","project":"openstack/puppet-mistral","branch":"master","topic":"constraints","hashtags":[],"change_id":"Id4fa0518ab7fb7d60f0d142666709d233db1966e","subject":"Update the constraints url","status":"MERGED","created":"2019-09-26 03:17:05.000000000","updated":"2019-09-27 04:21:57.000000000","submitted":"2019-09-27 04:20:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684903-1569558052881-91403195","meta_rev_id":"7ad85e6fbb37e3c1c40ed60d9afefb93bc08f7ce","_number":684903,"virtual_id_number":684903,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 04:20:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-26 04:24:28.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-27 01:48:33.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-27 01:48:33.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 04:24:28.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-27 01:48:33.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-27 04:20:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fec29fca0fb3e7d8879677d5078ecfb5aaee266b","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-26 03:17:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90b1684233ec5fed888fdf1fa3e30784ae6c6683","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-26 04:24:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/03/684903/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/fbed3bc/ : SUCCESS in 56m 22s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/03/684903/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/a41aff1/ : SUCCESS in 1h 05m 57s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/03/684903/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/e11d104/ : SUCCESS in 37m 47s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/03/684903/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/7f0c459/ : SUCCESS in 42m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c67fafb4d0d8770ecb590563c9e95ea72d8012b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 05:40:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd137495674a4643ad96883a70a80545 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65d3a3e5fd254f3eac2dcc9dfba5ea22 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/825a7f5338c3462c82b2dbb1f91d6e6b : SUCCESS in 3m 53s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4edddccaa6fd44fd92b8679b61198e2f : SUCCESS in 18m 24s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c59cb1ae8a2a4fb9aab3bdb2e9a5a80d : SUCCESS in 11m 41s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a838eb5c6b4c4a8d9948cf3a82869cf5 : SUCCESS in 10m 39s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9a60a9dc76d473d834b34d4ad92226f : FAILURE in 20m 03s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c7cecc12fa6547d5b18167e2897c980a : SUCCESS in 1h 12m 19s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8377e83257bc420f86f8e7e5c7eab49d : SUCCESS in 15m 20s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a81a8524613d48ce852edd832d8c3258 : FAILURE in 7m 05s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/15949457e5df4684b4bb12d4f8725275 : SUCCESS in 2h 18m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ec50a1cf44e4a8291530fffa5d1be79 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0ff5f415486ad75b37255b19d9079e2833403c19","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-27 01:48:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"080f286b7dca03bd5a02fb3a401100167a99b53b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:48:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d14b973124137ef0395c7d612d872ea47f17cb9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 04:20:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e50964c8739a4601a8cbc5df21cccdd2 : SUCCESS in 4m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f918f6474d4c4079b0367948fea9b97f : SUCCESS in 5m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f2a14eef15c41848e04a7ecd2b569a7 : SUCCESS in 5m 04s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2f26440bd7b0403db285bd9dc141ea99 : SUCCESS in 18m 00s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6740f9c9cd44ba3a19b6461b832c444 : SUCCESS in 13m 32s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9d9ea382cf9b4d23b417f1e5662dcdc2 : SUCCESS in 1h 04m 15s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6877fb281a40430eb802392900e209fd : SUCCESS in 24m 48s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/75413866a80d4e72a0104546907cf8b0 : SUCCESS in 2h 24m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d25dd830e608461a8ffe9524438b47f3 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8055a7d7a5384589748dbf2952790cbb9ea22f09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 04:20:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e287a7ba7ae30ed0d1c991c50a092ad271ba6439","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 04:21:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30450f2584fd41558cc8d1137c364fa5 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6fab895950f1fa790795988e9f8e3c307e679f25","revisions":{"6fab895950f1fa790795988e9f8e3c307e679f25":{"kind":"REWORK","_number":1,"created":"2019-09-26 03:17:05.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/03/684903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/03/684903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/03/684903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/03/684903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/03/684903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/03/684903/1"}}},"commit":{"parents":[{"commit":"610ddbd73f5c5e95a472e5811f41224a2ca879ba","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/610ddbd73f5c5e95a472e5811f41224a2ca879ba"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 03:17:13.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 03:17:46.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Id4fa0518ab7fb7d60f0d142666709d233db1966e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/6fab895950f1fa790795988e9f8e3c307e679f25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/6fab895950f1fa790795988e9f8e3c307e679f25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
