)]}'
{"id":"openstack%2Fpuppet-mistral~686667","triplet_id":"openstack%2Fpuppet-mistral~stable%2Ftrain~I146e25a226f87862f0aa390b098c73fe0dc3f8f9","project":"openstack/puppet-mistral","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"I146e25a226f87862f0aa390b098c73fe0dc3f8f9","subject":"Update .gitreview for stable/train","status":"MERGED","created":"2019-10-04 13:30:20.000000000","updated":"2019-10-08 22:52:22.000000000","submitted":"2019-10-08 22:52:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686667-1570575142502-0f5a3271","meta_rev_id":"2272a9a3fb4ad8490f8665391f4474f0854790cd","_number":686667,"virtual_id_number":686667,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 22:52:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2019-10-04 14:34:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-10-08 18:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-10-08 18:01:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 14:34:34.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-10-08 18:01:39.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-08 22:52:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a59aec0aa2026048a6143895a036df195ac45b99","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-10-04 13:30:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff7af8475504439cece56033dd0d1000e63e123f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-04 14:34:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5c13609c632b84020abddb550cfb339d36fc106","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-08 18:01:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d002d654507a92c6f546f67ac1b2f671a4fbe81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 20:36:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8983bb8eb9f149debc5faafc1308558e : SUCCESS in 4m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a35a53a03dc242d398c91885afe3df7a : SUCCESS in 3m 41s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7c5f360250504c10ab01aa1f8c055820 : SUCCESS in 13m 50s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd3631eefa414deda5e96b48d86ac357 : FAILURE in 23m 51s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7176ff914f0c43c5b94caa104213cd66 : SUCCESS in 1h 04m 06s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/904a5e63410842b39d4d8f51488aeb37 : SUCCESS in 16m 22s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25da843c919b44be815e5f52315385fe : FAILURE in 7m 38s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/cc807677e16147c69735284b4d209473 : SUCCESS in 2h 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1f9fc46c33ffe2564aae706f9eeb519a2f66926","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 20:36:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49062c30f5336df75121cc9fbd197ce63eeed72a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 22:52:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd0354a1c99a455bab9f79d24e48ed96 : SUCCESS in 5m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d48b051b199b44839ae7654a5034700e : SUCCESS in 5m 34s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/74a4da98f7674814ada088da4e9f1c35 : SUCCESS in 14m 56s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1e0fc63ab6204f6f99b2690ab05a93d7 : SUCCESS in 1h 00m 32s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2a0ce6df55304d8893b601e2adda21ce : SUCCESS in 16m 50s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c4038769386146eea893ddb83d80574a : SUCCESS in 2h 11m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4fe4e13aa9dd488827a64dd1f8c2a40eee4bd08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 22:52:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01db238764634596f926f0f1afee227df6e3880b","revisions":{"01db238764634596f926f0f1afee227df6e3880b":{"kind":"REWORK","_number":1,"created":"2019-10-04 13:30:20.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/67/686667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-mistral","ref":"refs/changes/67/686667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/67/686667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/67/686667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-mistral refs/changes/67/686667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-mistral refs/changes/67/686667/1"}}},"commit":{"parents":[{"commit":"d92cc8048b28eaaf358a628fb0cbf8955da5b350","subject":"Merge \"Update the constraints url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/d92cc8048b28eaaf358a628fb0cbf8955da5b350"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-04 13:30:16.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-04 13:30:16.000000000","tz":0},"subject":"Update .gitreview for stable/train","message":"Update .gitreview for stable/train\n\nChange-Id: I146e25a226f87862f0aa390b098c73fe0dc3f8f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/01db238764634596f926f0f1afee227df6e3880b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-mistral/commit/01db238764634596f926f0f1afee227df6e3880b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
